)]}'
{"id":"openstack%2Foctavia~832603","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I4a9d14f0f918bda64ec489777ef7c10ceeb9b96f","project":"openstack/octavia","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-03-15 17:21:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4a9d14f0f918bda64ec489777ef7c10ceeb9b96f","subject":"Fix Amphora RSyslog configuration for TCP failover","status":"MERGED","created":"2022-03-08 12:21:16.000000000","updated":"2022-03-15 17:23:42.000000000","submitted":"2022-03-15 17:21:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832603","meta_rev_id":"48c83f38cf00ea93b2d8c8a1c06f945035db5e9f","_number":832603,"virtual_id_number":832603,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-15 17:21:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2022-03-09 09:38:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-03-15 15:08:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-03-15 15:08:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-08 14:57:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-09 09:38:16.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-03-15 15:08:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"99992daf160bf5817a71b25222ecb6d2e3854c78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-08 12:21:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe54d3a33fdb7c64d8c42983929b43b1460b4594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 14:57:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f12c34908afc4cc8a66c6e6bad6f7fa6 : SUCCESS in 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b2d25853014464983fc2f99735e1f87 : SUCCESS in 8m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b43117f9dadf42b4b9ca63a0800990d5 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5edbca499ec3492581e83eadb21154b4 : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84319dafab884edfbdee17622c15b7ef : SUCCESS in 13m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a72e4a628b7c4707868b639521c8de7f : SUCCESS in 6m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/338ff9116b7548eebaef154fed660ff3 : SUCCESS in 8m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/eff19782192040a5bf08f3b13cb8ab0d : SUCCESS in 10m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f20310d2167643d39aa1ee52566fcd6e : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef0d785dd9e246d887fc2b4b445a74b5 : SUCCESS in 9m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d83889d1167c49dfb2cab188343fc1f6 : SUCCESS in 1h 30m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e58e441ad72470cb3ec702e7862d099 : FAILURE in 2h 02m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6515dba8e0784bfaa8b089e5c982d7d6 : SUCCESS in 40m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/35ed8d90badb45ff989bc55c4ce7661b : SUCCESS in 35m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5eb3bf912b804e00aa08c1b80d7c5c2f : SUCCESS in 36m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5881a455fd33471094bf0dcdd296a2ee : SUCCESS in 52m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d819303ee6b64bb4b8f410b6edaad6eb : SUCCESS in 1h 36m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b442bebf22ac4c6c928548b235467977 : FAILURE in 31m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b9e8d987fbc14c3897bdea9633d28618 : SUCCESS in 1h 33m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/08404f49c2dd418199331ddc60714a69 : FAILURE in 1h 52m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f831102973f247369d78563fcd55e34d : FAILURE in 1h 25m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5cb95938b9104b2fbd4e102eb0109acf : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e4f308b0aaca532919b45536f79465628765054","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-08 15:04:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"865268dfac80799d03ec15c175c4eaa6b86bb7f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 17:19:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8392e0bf836f40bea771aab0fd8b97f2 : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a1ffa03b5ee4ee98dd9da79bdc3cd57 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/633b0acc37994d9d8e7bd69ce8e9f217 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f7c0df5bfd246399eaeed20880a0ece : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c321ca6bf76f4e30bdaff2487d022fcc : SUCCESS in 14m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b7b806c6d0d47eb9fdb103c923f616f : SUCCESS in 6m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e5a1bc8981324b61a07baa716f3d7f34 : SUCCESS in 7m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c5bc6926ea0b46cd9746ec039c61df9e : SUCCESS in 9m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4c41f0434a0a49a992e0b2fe6b8fc3c4 : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2f6f5126dbed4b2f917f0258e27be844 : SUCCESS in 9m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a9fe830746654c248267f40ff9149dac : SUCCESS in 1h 31m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/24fca4b7ce654967b30ab99f0009ad72 : SUCCESS in 1h 59m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f29485c5e9d24c3ea8f3c96207da42b9 : SUCCESS in 47m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/dc19ca300a8142db9d757c3b274c801b : SUCCESS in 37m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80ebd6b33f11470f9b3d6d7f0ad51265 : SUCCESS in 34m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/952022d0c7f247bd91a244369c7831f1 : SUCCESS in 51m 48s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5c8fda11a7384b4b8afaaeb42d339259 : SUCCESS in 1h 32m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0ebb1d1c23384b3998b4e5476de10978 : FAILURE in 34m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6bacbd28fe794bfeb059de4f8d2e5759 : SUCCESS in 2h 01m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6bd572fef4194756a82ee44c6a9e5ccb : FAILURE in 1h 33m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d3f229c7660f425693fd0b8220260f62 : FAILURE in 2h 10m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fe33fd1734924b14927f1cdc515a0936 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42b715fbc49533c2d8e748b61cfede3f660e7fb4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-09 09:38:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"250446aaf1a59e41a83ce450b15254857e829dc4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-03-15 15:08:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea5e41f75493198e987de5137dd6a45604c6f3ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 15:13:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bff2599b3ed172e23e38c2a229ccbc97559fda11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:21:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0deb3431da924f8a8fa54828eb8873f7 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c70744cdb6df45b2bee4b59ba751ce42 : SUCCESS in 8m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/208d5675e6e74990b8c7c38c47a414e5 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ca80f0f14fa48c8bd6ac2adef7c7a08 : SUCCESS in 14m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fd960785ff6402791dacb9e9df93ff7 : SUCCESS in 7m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2d8da50107624eb0a8d46c2b69d4990a : SUCCESS in 8m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/08e9353a41e240d5b2d3ab082b82e122 : SUCCESS in 9m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/96e72765588f4320ae1a12597752274d : SUCCESS in 1h 34m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7a2ff5ef6964e05ae74217823455526 : SUCCESS in 1h 45m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cc0677fad9564631bfe5f54cbd3a6937 : SUCCESS in 54m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1391e87c664d42c7b75653e4b5b7d30f : SUCCESS in 47m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98e21d32c74246a09ccf3ef2b72795fa : SUCCESS in 52m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4e9c9d51a2a14238bdf884201a843b8e : SUCCESS in 54m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4241087ac92492d5d692820da5e4488e95a8c277","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:21:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"48c83f38cf00ea93b2d8c8a1c06f945035db5e9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:23:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec907f06f3c4f4a97bd9ac009e555d4 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96f2530a1d4340bb89c40496cc1e1f49 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5811fe88c0fe70949e5a75befd2456ca524accc","revisions":{"a5811fe88c0fe70949e5a75befd2456ca524accc":{"kind":"REWORK","_number":1,"created":"2022-03-08 12:21:16.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/03/832603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/832603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/832603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/832603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/832603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/832603/1"}}},"commit":{"parents":[{"commit":"64c792c225a510f8c27a5aa27ea5dae261e0b76c","subject":"Merge \"Fix nft command line with negative priority values\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64c792c225a510f8c27a5aa27ea5dae261e0b76c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-02-28 12:13:08.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-08 12:21:05.000000000","tz":60},"subject":"Fix Amphora RSyslog configuration for TCP failover","message":"Fix Amphora RSyslog configuration for TCP failover\n\nRSyslog failover does not work if queues are defined on the actions. A\nworkaround that allows to use queues with multiple RSyslog servers over\nTCP is to combine all actions in a single ruleset.\n\nConflicts:\n\toctavia/common/jinja/logging/templates/10-rsyslog.conf.template\n\nStory: 2009876\nTask: 44599\nChange-Id: I4a9d14f0f918bda64ec489777ef7c10ceeb9b96f\n(cherry picked from commit 145f4f804347358689165d1418756fe2a91c07da)\n(cherry picked from commit 5ade96be53f2fd75a49964d9c7ea98f684ec970e)\n(cherry picked from commit 3f134a292f907c2eb8e05aa85b80e7a28c8b44ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5811fe88c0fe70949e5a75befd2456ca524accc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5811fe88c0fe70949e5a75befd2456ca524accc"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
