)]}'
{"id":"openstack%2Foctavia~832604","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I4a9d14f0f918bda64ec489777ef7c10ceeb9b96f","project":"openstack/octavia","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-03-15 17:21:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4a9d14f0f918bda64ec489777ef7c10ceeb9b96f","subject":"Fix Amphora RSyslog configuration for TCP failover","status":"MERGED","created":"2022-03-08 12:22:49.000000000","updated":"2022-03-15 17:23:43.000000000","submitted":"2022-03-15 17:21:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832604","meta_rev_id":"6d89c10cdbd52fb33c0861fda5b0a7f80fbd0494","_number":832604,"virtual_id_number":832604,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-15 17:21:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2022-03-09 09:38:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-03-15 15:09:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-03-15 15:09:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-08 14:26:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-09 09:38:28.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-03-15 15:09:03.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"b229e0db3e6dd898342b0f351dfab2eed00aac2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-08 12:22:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"719b35c3bcb10449bdee2f386a5b072d2f786398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 14:26:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf680c2713664bb396cc4222cad5d146 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ada82fff0fa04ba091011ac4fb2d8b1c : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44823f8c14364ecfac972077732f026d : SUCCESS in 7m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/028e3ad457b34531acb1d38a9790cfb6 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ea435ff7f64b4ba7ab3514c219dc73 : SUCCESS in 14m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bde6c58de9bd46a5b18563b36ca7057f : SUCCESS in 7m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ab8a3e83753947bd8d914e7c046fc983 : SUCCESS in 8m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bfa136606e4e42d2993a70f2ac8fa8cf : SUCCESS in 10m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4eceb7b0a83247bc8c2aed8c9d04b1ff : SUCCESS in 9m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ba1cbd936a5b467d97e3e2951b411077 : SUCCESS in 1h 08m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b482c12eea2645668b0da58741a8119f : SUCCESS in 1h 40m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/47de4699823b4826956093f23b2314e6 : SUCCESS in 36m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/aea8a89294ad4d1c9ade9706d0ecdfdb : SUCCESS in 35m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bd0f817b98e2483db932bd36734814fa : SUCCESS in 36m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b3346c03e3c24c71be76de1c0f705015 : SUCCESS in 54m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/18324b47d3b54356b766f96d54c3cd1f : SUCCESS in 1h 25m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ea1d8dee6c6437c8806c619034c80b4 : RETRY_LIMIT in 8m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5f337a7380ad4e2a84d58c4530eb0da9 : SUCCESS in 1h 46m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f50af98f1ab22342901d9c8222a2421d49e7bdd6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-09 09:38:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9958e0c08c19b68fea838c548e39422b949e2bbd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-03-15 15:09:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e847495041ec9b030aef599efe7722645a1b56b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 15:13:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35fd4c2a2c97ee2b641cc6bc7a31c4f6c7d2d39f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:21:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b7c467c73fd4a8e85526c8e22045dea : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d33487c7800e4bf59d509a2ffebe2309 : SUCCESS in 7m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da179e6ab94441ebb6b10f1a4f315b61 : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/619d0e6af2324180a6e610a601a9fada : SUCCESS in 14m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/befc74fcaaa845c7990bbb7b83ba41e0 : SUCCESS in 7m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/360c22860eb34bba85f2df721d94f6bd : SUCCESS in 9m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/04020525da14467f9ac38c04977cf629 : SUCCESS in 1h 25m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a20c03d9884404eb7846cada5a913df : SUCCESS in 1h 55m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/69ec0dd0b23241d7bea86cc964881628 : SUCCESS in 45m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cbad4e834f5d49ea8dc305f2594969da : SUCCESS in 38m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e016581cbaa449f9b7a38af0eaa9973 : SUCCESS in 50m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/53a7a29f3c784db8b00fbe1f6871c889 : SUCCESS in 1h 08m 41s","accounts_in_message":[],"_revision_number":1},{"id":"1937689fd6aa64227981da33750d889667668b42","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:21:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6d89c10cdbd52fb33c0861fda5b0a7f80fbd0494","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:23:43.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce6a07b64ec74330b25ec4263f4b33dc : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6c6f0facaf84b72993f1783ecf7f961 : FAILURE in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d63bcf507716b2370165999da4c63dbf9f4208c0","revisions":{"d63bcf507716b2370165999da4c63dbf9f4208c0":{"kind":"REWORK","_number":1,"created":"2022-03-08 12:22:49.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/04/832604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/04/832604/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/832604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/832604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/832604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/04/832604/1"}}},"commit":{"parents":[{"commit":"1d8ac7e2c13a1ba5370eb8708f5eb83f72d15a5c","subject":"Preserve haproxy server states during reloads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d8ac7e2c13a1ba5370eb8708f5eb83f72d15a5c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-02-28 12:13:08.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-08 12:22:12.000000000","tz":60},"subject":"Fix Amphora RSyslog configuration for TCP failover","message":"Fix Amphora RSyslog configuration for TCP failover\n\nRSyslog failover does not work if queues are defined on the actions. A\nworkaround that allows to use queues with multiple RSyslog servers over\nTCP is to combine all actions in a single ruleset.\n\nConflicts:\n\toctavia/common/jinja/logging/templates/10-rsyslog.conf.template\n\nStory: 2009876\nTask: 44599\nChange-Id: I4a9d14f0f918bda64ec489777ef7c10ceeb9b96f\n(cherry picked from commit 145f4f804347358689165d1418756fe2a91c07da)\n(cherry picked from commit 5ade96be53f2fd75a49964d9c7ea98f684ec970e)\n(cherry picked from commit 3f134a292f907c2eb8e05aa85b80e7a28c8b44ec)\n(cherry picked from commit a5811fe88c0fe70949e5a75befd2456ca524accc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d63bcf507716b2370165999da4c63dbf9f4208c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d63bcf507716b2370165999da4c63dbf9f4208c0"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
