)]}'
{"id":"openstack%2Foctavia~834636","triplet_id":"openstack%2Foctavia~master~I0f4c59a2928f2a813171109aaf73a1d7dff9eefe","project":"openstack/octavia","branch":"master","topic":"agent-remote-execution","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-04-08 15:29:17.000000000","reason":"Michael Johnson replied on the change"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-05-05 14:59:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0f4c59a2928f2a813171109aaf73a1d7dff9eefe","subject":"Reject invalid whitespace in HM url_path value","status":"MERGED","created":"2022-03-22 09:34:51.000000000","updated":"2022-05-05 15:00:43.000000000","submitted":"2022-05-05 14:59:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834636-agent-remote-execution","meta_rev_id":"b030676f7931c829b2963d6bc9376e5ee931bdcf","_number":834636,"virtual_id_number":834636,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-05 14:59:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-08 15:29:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-05-05 12:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-05-05 12:54:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-05-05 12:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-22 12:16:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-29 22:29:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-05-05 12:54:29.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"b4d93d66a2579a48a168f2499f69e5efdde999e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-22 09:34:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1835128d714de1b11d6b0723fb35a6a63ad8ca14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:16:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed6244c4c5744b6dac6e80e40db79faa : SUCCESS in 13m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca60bbf706654440895254c3831f5d5a : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69e67fda1f0f42448e5acbabb5042086 : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/036e4b78a747461a8eb15f56080f7a03 : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88451959e5984f75b75fca2c0e5c10a1 : SUCCESS in 8m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08315f03443846c4bc2d7a0d65c28ce5 : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebf7f8a44be2436ba28a8a41618d13f4 : SUCCESS in 13m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d7967e165cd4140976f8c90b5c4141b : SUCCESS in 7m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/baf48f5062b8465fa7007be9d9dcdb8d : SUCCESS in 7m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4d1b73d0201646178ea8a49477e5a973 : SUCCESS in 12m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/37419261b7c14bc2b9755c9d6757307a : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/226118ce3ad94e6e834b417199adb930 : SUCCESS in 11m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b03dd575f99d434e8885c68c0690d382 : SUCCESS in 2h 10m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/10ea0ffd307342d780d25bbd30eebc4f : SUCCESS in 2h 02m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c52f2da1515440b7bd9d530f72dd19c8 : SUCCESS in 45m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9e543f1888a7449ea9857d4fab34e77a : SUCCESS in 1h 21m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/51d5fd59e789455d9e776c6c9807911e : SUCCESS in 1h 25m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0788f6803d1b4c86ba2335bb033c4b58 : SUCCESS in 55m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8bb2429f90f44fe4a4f3ba5a0d8aac19 : SUCCESS in 1h 38m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ddd933fa70554ab1b2d55d489a402bc7 : SUCCESS in 1h 49m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4008f1261e2a478f9bd1f3650b68a717 : FAILURE in 20m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5ec93ac603f94a3e92e363a324123456 : TIMED_OUT in 2h 38m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6dfe9668ca5c436aa33c259d7e46c454 : SUCCESS in 1h 25m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/8a08e900bf1d4cb883d761585ddd0ea8 : SUCCESS in 2h 10m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1dd32b9c642aa2df2eade167846e1185230563dc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-03-29 22:29:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8c22053038c7db1050a3897b52f6426fc2c75975","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-30 09:33:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"64a1d1436589cdd498eb8ce9002bd9a9741a5ab1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-30 09:35:49.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b1f878081ee33321ab7cc87fa8d4afa3b58fe5db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-30 09:42:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c1a47c5524d77f529043d3213f9fe3c19d76285d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 11:52:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea5acd3db581417c8615960deaa88783 : SUCCESS in 13m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45016be9d1d84198a06d8297a730c17f : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e55f2f7d30014de49ae424006d99f4da : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7287afb36d7041638b5ef5f185f947f4 : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c00265b6c36241f398f0457bbcc0b5b7 : SUCCESS in 6m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eedb95cbf5454f63b3f59b5d6846e652 : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34d55d4f29074a93896c62a40ff7d799 : SUCCESS in 13m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e320d8a8fcc1481183b436dba1116902 : SUCCESS in 6m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/09180494f92c46739c707fb91643dd41 : SUCCESS in 6m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/22d80b752a1c4532b89fc3a98e1a3182 : SUCCESS in 11m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5b35d3351c06406ab3dff4c50b138e4e : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dc54f1f013c44b7495341ab213a538b5 : SUCCESS in 9m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1592299c1f994315a0b27b5626d801ba : SUCCESS in 1h 45m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9886413ec31549ba97076a6ed99a6bd6 : SUCCESS in 1h 41m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1233ce51b3e64ee2ba635596074a073d : SUCCESS in 38m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5c1b2bdbde6d4c37b301dbea267db276 : SUCCESS in 1h 02m 37s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/66d4c1219a0b46f797e83c3ad39b0dcd : SUCCESS in 59m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d49f48e63a1c4420a706efa29111b8c7 : SUCCESS in 36m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1e77221f70ca42089577d7d76fbcc17c : SUCCESS in 1h 26m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9c694f190a9242d1ace7425447a2daaa : SUCCESS in 1h 40m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e5555d5295bb4a37a010ed80a51062a6 : SUCCESS in 1h 51m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ac0dc9c8df544f439a22d9519708772d : SUCCESS in 1h 30m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/30075bef8692433ca6392b3a97b48423 : SUCCESS in 2h 00m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4978661a87249bab10cc9c1f883a60e : SUCCESS in 2h 04m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cab052655c26e7b77ab8317fdde8f5ad0343aebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-08 13:00:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1624c7f7c9df72e6e4ff147495ea3fce30be89ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 14:59:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe5284506262410c88862670a0cc0474 : SUCCESS in 13m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df23041c3fb648409b9d81e5f9150e89 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e53c6919b19b4c708da21bb8aa6d88e2 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf95958be9224176ae633f3a2a0c0263 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f5a18d718d741b5ab01653abc9e650c : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e11d49c6741a4967bb6709e0dc143251 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d1ada3385874c3cb485b3f3433ac582 : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00eb96ba36284c3696e0d05b191c81e8 : SUCCESS in 5m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8494eaeba93f45039d577477eff2d110 : SUCCESS in 5m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/077f1a78f5a14cc1bcee363d7f1ebc96 : SUCCESS in 8m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6a02ce44744945b3a2aabe53a4ee6ae4 : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/90cd6afac15a4c91ac27b44eadd02624 : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b2549ad391c241b4aefc46b8dd305f5f : SUCCESS in 1h 57m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a55c35f91445408cb58d1acfd5bfdf21 : SUCCESS in 1h 20m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d7e504d40aa94177b27096c16bf1d4f2 : SUCCESS in 39m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5b83fce3576d4b9eaa813142e2fcf931 : SUCCESS in 51m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a07fd557a9ad484bac7badd91fc9b106 : SUCCESS in 57m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f69f8e01884543e2b8a43df90a7f870a : SUCCESS in 36m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/27568e98e3c64858a78d7dd5d66c0135 : SUCCESS in 1h 53m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/270cc5f9dce842209e807ab27e3ff392 : SUCCESS in 1h 46m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0d7dd36d89424ef28bc71fd15a5c9336 : SUCCESS in 1h 26m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d9655b2f2a494f788e429fa639571c7a : SUCCESS in 1h 31m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/348ea0e36dd944f5afadc3003645525a : SUCCESS in 1h 24m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a8adcccbe4b4e7e878d8811d7d06c1c : SUCCESS in 1h 57m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7b0a6feb84e00a50bcb96d91019519eb86ef53f3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-08 15:29:17.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b36e9168a97c47427e2d8b2d763168c2df6377de","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-05 12:54:29.000000000","message":"Patch Set 4: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2117697088c30eeddd751a0922ad2ce55d005ed3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 12:56:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"42ed9310749f04807683193f9a5ea48997f9e97b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 14:59:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4587806b10442808ab5aaff42316930 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/180e95ec6d3a482c9118237b985140b7 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eac74c6e4036422381d439581fbe668e : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eaf3a21246fa42d88a9c25793a230bd8 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6583f8477b648d8bf2ac01ebd4425bd : SUCCESS in 12m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ce09dde80e745daafad01eeef9384e5 : SUCCESS in 5m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4000c09b630f4962b2f498440c473cb3 : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/68426bfb1af34330b63603ac02a5e928 : SUCCESS in 8m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/25543a3bc6ab4a70adacf4b3b2e47d93 : SUCCESS in 2h 00m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/817491d1d22b40d29ad8b755835ef1fd : SUCCESS in 1h 38m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/04a3d4a43f1f4a6b9290856bb6eca322 : SUCCESS in 44m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0cd5657b0a8146948f4012450a19290f : SUCCESS in 1h 10m 38s","accounts_in_message":[],"_revision_number":4},{"id":"5ea6b4855017bd410caa8420e3f0be0ecefb515a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 14:59:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b030676f7931c829b2963d6bc9376e5ee931bdcf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 15:00:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2135247d205e4da2aa089ebbc51005f5 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4ba7d61014d4a6b92e80956ade38375 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"251fab17b51f99f1f0c1cce4985693993d8fb957","revisions":{"a678edc0bdf75884c6d3420a37854ae4580ea8cc":{"kind":"REWORK","_number":1,"created":"2022-03-22 09:34:51.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/36/834636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/834636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/834636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/834636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/834636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/834636/1"}}},"commit":{"parents":[{"commit":"83ae4a6335a78f8388ec7cde481c729dff158d4c","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/83ae4a6335a78f8388ec7cde481c729dff158d4c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-22 09:02:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-22 09:25:29.000000000","tz":60},"subject":"Reject invalid whitespace in HM url_path value","message":"Reject invalid whitespace in HM url_path value\n\nPlain whitespace characters in url_path must be rejected because\nthey may allow injection of arbitrary HAProxy directives in the\nconfiguration of amphorae. This prevents a remote code execution\nvulnerability.\n\nStory: 2008994\nTask: 42656\nChange-Id: I0f4c59a2928f2a813171109aaf73a1d7dff9eefe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a678edc0bdf75884c6d3420a37854ae4580ea8cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a678edc0bdf75884c6d3420a37854ae4580ea8cc"}]},"branch":"refs/heads/master"},"1a9b2c71500335a09ecbfaff9b27c5f9ced2543c":{"kind":"REWORK","_number":2,"created":"2022-03-30 09:33:56.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/36/834636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/834636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/834636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/834636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/834636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/834636/2"}}},"commit":{"parents":[{"commit":"83ae4a6335a78f8388ec7cde481c729dff158d4c","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/83ae4a6335a78f8388ec7cde481c729dff158d4c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-22 09:02:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-30 09:18:27.000000000","tz":120},"subject":"Reject invalid whitespace in HM url_path value","message":"Reject invalid whitespace in HM url_path value\n\nPlain whitespace characters in url_path must be rejected because\nthey may allow authorized users to inject arbitrary HAProxy\ndirectives in the configuration of amphorae.\n\nStory: 2008994\nTask: 42656\nChange-Id: I0f4c59a2928f2a813171109aaf73a1d7dff9eefe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a9b2c71500335a09ecbfaff9b27c5f9ced2543c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a9b2c71500335a09ecbfaff9b27c5f9ced2543c"}]},"branch":"refs/heads/master"},"ed876582731f96b320ae6ce6b3fdd5a0465de4c8":{"kind":"REWORK","_number":3,"created":"2022-03-30 09:42:16.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/36/834636/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/834636/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/834636/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/834636/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/834636/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/834636/3"}}},"commit":{"parents":[{"commit":"83ae4a6335a78f8388ec7cde481c729dff158d4c","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/83ae4a6335a78f8388ec7cde481c729dff158d4c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-22 09:02:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-30 09:37:51.000000000","tz":120},"subject":"Reject invalid whitespace in HM url_path value","message":"Reject invalid whitespace in HM url_path value\n\nPlain whitespace characters in url_path must be rejected because\nthey may allow authorized users to inject arbitrary HAProxy\ndirectives in the configuration of amphorae.\n\nStory: 2008994\nTask: 42656\nChange-Id: I0f4c59a2928f2a813171109aaf73a1d7dff9eefe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed876582731f96b320ae6ce6b3fdd5a0465de4c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed876582731f96b320ae6ce6b3fdd5a0465de4c8"}]},"branch":"refs/heads/master"},"251fab17b51f99f1f0c1cce4985693993d8fb957":{"kind":"REWORK","_number":4,"created":"2022-04-08 13:00:25.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/36/834636/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/834636/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/834636/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/834636/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/834636/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/834636/4"}}},"commit":{"parents":[{"commit":"83ae4a6335a78f8388ec7cde481c729dff158d4c","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/83ae4a6335a78f8388ec7cde481c729dff158d4c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-22 09:02:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-08 12:59:39.000000000","tz":120},"subject":"Reject invalid whitespace in HM url_path value","message":"Reject invalid whitespace in HM url_path value\n\nPlain whitespace characters in url_path must be rejected because\nthey may allow authorized users to inject arbitrary HAProxy\ndirectives in the configuration of amphorae.\n\nStory: 2008994\nTask: 42656\nChange-Id: I0f4c59a2928f2a813171109aaf73a1d7dff9eefe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/251fab17b51f99f1f0c1cce4985693993d8fb957"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/251fab17b51f99f1f0c1cce4985693993d8fb957"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
