)]}'
{"id":"openstack%2Foctavia~835616","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I48a02780adf54b8d6d3bedada448a4f91e0539ed","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-04-27 12:57:14.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"I48a02780adf54b8d6d3bedada448a4f91e0539ed","subject":"Pass timeout_dict to _get_haproxy_versions","status":"MERGED","created":"2022-03-29 12:13:14.000000000","updated":"2022-04-27 16:49:13.000000000","submitted":"2022-04-27 16:47:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835616","meta_rev_id":"09558e5668b5ec1cc1ecd5f06cb59c330cc1bc16","_number":835616,"virtual_id_number":835616,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-27 16:47:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-08 12:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-04-14 22:47:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-04-14 22:47:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-29 13:56:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-08 12:24:14.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-04-14 22:47:35.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"59e1d2bf6eab43db58b329456b42166d2083dee8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-29 12:13:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00a7153dc8c57e560552b135aacbd834a42d9540","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 13:56:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/208ae0a9d32043ada5986bcc9fbab8ca : SUCCESS in 12m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/583cbf7b07dd47f19479ab199949b37f : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c1f255c8ab64ad0843521a72c1673f8 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45922bf1eb7d40b0990a2b54438b9570 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a447b49268443338645b6dd8efe4e6d : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/666c94f81000411ebcac68a3ba066727 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa9aeb809c704171a9e5898ad9f3eaf6 : SUCCESS in 13m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/965e460310e1441ab6f1cefc8bfae13f : SUCCESS in 6m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fb6cef9f1ab046b185055d0d11307e7b : SUCCESS in 7m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0e0fb547ea454b269e1f6792de729325 : SUCCESS in 9m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/431468fdf5f04592bf53c481026d77d2 : SUCCESS in 7m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6dce5cf70dc644faa0634db7ad19cfd3 : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/23a452cc8fe0434386a44a7a6e6e680b : SUCCESS in 1h 30m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/386205b14bfe4250b9fc99d29be8cc97 : SUCCESS in 1h 25m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a82af842f2ee458e8502b52d8529dfb2 : SUCCESS in 35m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/169d6579fc194b16aa71f13f910ed6b3 : SUCCESS in 33m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/32475be7a9b3400d81db29ec374bd10f : SUCCESS in 35m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/64e0101d4d2c4df590be4f5c1d6b31f6 : SUCCESS in 49m 35s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/35226ea7860d4387bb17b80da8716eb4 : SUCCESS in 1h 19m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e10729615dfd4bfc944f4cc58ef54e3d : FAILURE in 30m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/db28ee599c844f95a76f7cc3a98d9f8f : SUCCESS in 1h 40m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5fd300753704431bb84b6d2fd8a2a130 : SUCCESS in 1h 33m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/092686bde3334277adf89dfbbcb42aae : SUCCESS in 1h 26m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9d6018970bba4622a9733376b93ccb89 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4022cb5efc100395f9ae63144a46ef7012050e30","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-08 12:24:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1292c28538b382c6e7899568e1775e23830364b1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-14 22:47:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"05701706178b8a8acb62eb9088d42986d7b10e13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 22:48:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7ca2c8a1971395886ff72404211637d9cfb50ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 01:12:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b47da4348c747519168e6f1a5314f80 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eabad3385a54974b576a2eca6fe5d6f : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8dd07134e1594233b6df09d0313c7011 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0537d360ca884c07aba1491394a627d6 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2a7fc3364f248d8b8233dedb5f9c98d : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c50a534a9664f75a851c582d98b96c2 : SUCCESS in 4m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8dff18d3b0de44d8919b38c444ef44c7 : SUCCESS in 5m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9b09ed0ad20c410aba84c6dceb8209dd : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0dcaa3ce58dd44bcac0f7833effc78c6 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7442b42467774666ad1c83f86bab1aa2 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/85460989262c45d492afcc55f2e5a018 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/818183780b1640b8afcb136d11318873 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18b843ffeca244c0ad23ad1745e0d34e : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/401eb089da9540aab191e1a0927e4ec8 : FAILURE in 10m 11s","accounts_in_message":[],"_revision_number":1},{"id":"08bfd69fa39fae1b6be31a273b4af41199c4b3d4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-15 01:59:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42472f35a5d055e02c36f7f6aa1b99e871bbe0c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 04:44:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e66025848c84476bfafc89c8a7aeeb6 : SUCCESS in 9m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8a2c197e969465b993b8afebbecc9d7 : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a87b3c163844a5b820e9175f463b9bd : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3e67ba8466348a9bf704c59d144cb8d : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70427d48c0844b9885034457a46df5ff : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa6eeaf4e02346c4a27c9883ee9c85cd : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b151cfb89c7476ab1e0b32791b6544f : SUCCESS in 12m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c018531ba6c479f80dcaf80fa99064d : SUCCESS in 4m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/aee74133ef664adba14afd322c1e4390 : SUCCESS in 5m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7565f3d95acc4cffb35ee4bb28e9fe86 : SUCCESS in 7m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/16275b8e175344eaa6703bd25c2a66bf : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ac6c00365f9544ab95fea4d2d6dbc157 : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/65c2e189274b4b5aa8ab3e03ddf78d7d : SUCCESS in 2h 35m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2f20114dfad642278a08dea79492035c : SUCCESS in 1h 42m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/97f89b3feec047009ab0d9234ed149f5 : SUCCESS in 40m 40s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/943f608ad0ac4f788d2bf604134bb7d4 : SUCCESS in 34m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11351765d9ec4d03bd0cfaf287d1e923 : SUCCESS in 34m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/28d0554dbb9b4829967990438eb2fa30 : FAILURE in 11m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/28a164fd53a744fb9464503cb561e02d : FAILURE in 51m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7cc6562bb9e2497aae18922955a749bf : SUCCESS in 2h 01m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e74c21dde8b340e9b127a2bee526c355 : FAILURE in 30m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f055715f633c48bfa4f7e3eb72894023 : SUCCESS in 1h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b8a4863be282407c9d03ce73a1c8bf80 : SUCCESS in 1h 45m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e3e4a785b6b54ed1aa1a5cda518455c1 : SUCCESS in 1h 23m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fe9d9d634a064dcfbb86c9b2a63a71bb : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf508e90e76198ce9affbd5867255627398de5ef","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-19 16:04:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"005c49c2d45fca7a0f4e115154991d6d3e6c7949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 17:44:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca2c0849f6984782b0ff04c48b97b4e6 : SUCCESS in 11m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8333a26a1c254816bd81f0ed4f5fb3d0 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ffac8c66d8244eda184aebc83f1ae22 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5624ce4a5dcf410db05470c546f6581b : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91280b2924344c3ba1782ca817a3c79d : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b625362e8dfa433e93a0e9119a9c40f8 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3522c899cef747fdba78ba79090142eb : SUCCESS in 15m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e182673f49b349d7a18945c996766194 : SUCCESS in 5m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bec220aeaf0844de9abcd053ff5e28b5 : SUCCESS in 5m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b0aca08bf1b443d1b05f87a16ff91eec : SUCCESS in 9m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fe889cae2d6c4f3a94a4b6aaee47d0e3 : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4468daab46984a2e8a7c2f3e54550792 : SUCCESS in 8m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/23fba80470ac419db56ba4b506f2fb6d : SUCCESS in 1h 34m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46c0c554188d4ddbaaba4ee53b287a82 : SUCCESS in 1h 20m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8142cef83fde4e7e9994a4aaf554b11c : SUCCESS in 38m 10s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5b371191727a4a348889c83452500108 : SUCCESS in 32m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/186bbbfa03634cfb8fe4d9ccbc9b1502 : SUCCESS in 35m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3d31123688194c4c9067411a1ea6df7b : FAILURE in 24m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/401abc10c54d48f68dbf97a43e622bda : FAILURE in 41m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/eb7c3aef0c734cc59cb6be3be4b52c38 : SUCCESS in 1h 27m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/439e9d0aad5940b7bf09c813cf3ab761 : FAILURE in 31m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/73ca099b343f4ed3b57b22be82a06a84 : SUCCESS in 1h 36m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0c74be5e259145dcab8124194c9a5b43 : SUCCESS in 1h 34m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/98be9dffbb0441cfb1583e09015488cc : SUCCESS in 1h 16m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1ac039f8044b4f468dc49e3f932efbf0 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1dfdebb04592602cf22418db256a3b3ba2021cb8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 12:57:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3a49de515eb7492236d174b1a864e875f043a69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:08:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b2228cf2d144591b74979b470ba154f : SUCCESS in 9m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e72751e933ef432592682c2254b66ce0 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/233b24aaad34491dab045f6c3fe3f715 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc081ebc11dd49059764d6d3769faa5a : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f14f0d1ea2249359f9e4d3d8499d776 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b55c25b137f947c4846be62d5ffb6247 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83e08da082794002a87a48015a9abe89 : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1545babe42c4ec3821e7150029a9b9e : SUCCESS in 5m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5914aa1dc45e499cb2f120ee64cd722d : SUCCESS in 6m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2a5f6957c48946ff81f9c195dd9b913c : SUCCESS in 9m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ab78fb1183d745b19ebb8f4e0aba44ef : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/067463ec1c1749afa8d521da6972fcfc : SUCCESS in 7m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/95b29ecd178345d58285991f91e8e6ef : SUCCESS in 1h 30m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f34013e94a34519a6bd8437f53f48a2 : SUCCESS in 2h 04m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a3224d986f0b4a539ede354d496f662e : SUCCESS in 35m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/54bf04ab8ac3438798b304159f672510 : SUCCESS in 35m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a61f40092c44b6b80104e4cbee2cfde : SUCCESS in 43m 10s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3980efe434e845c590e5f28595046df7 : FAILURE in 40m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c688daa622604f658b1ac029274b240a : SUCCESS in 1h 20m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/31988ab4ce2741779fcf50258c823531 : FAILURE in 29m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/df63469dd4c04c1d9fab1226752d5f15 : SUCCESS in 1h 24m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/045f61771240434d8e6337cbbd97a522 : SUCCESS in 1h 45m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8f981de0f1324550b1bb396b6712bdc6 : FAILURE in 1h 17m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/dee901b8a2094253b3fe7926b2467e3b : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e170c5662b9eff8420fad194807376a056b53042","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:09:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c618797c47a18ffc9ffbad09b5a59306980f7034","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:47:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1e7e19f6e574840b58fef173fd06e5e : SUCCESS in 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4852be4f5f84e048585103b60d43d14 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3556daf0ff874c078c5c4a6ca809fbec : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f1c4f816ff444b8893346191246608f : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c198e94592a34b678daf6493a8bfd234 : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9fd98a20b6f485e9f263bfd8d9f5932 : SUCCESS in 4m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3dfcc6270fc445268ece229022134e39 : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b4f36d4ed6a74114a8e4147d1778814c : SUCCESS in 6m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eaa86bb745e9441a95710371a05e2f81 : SUCCESS in 1h 16m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8aab0e42af58495db7c1ae7f5364164a : SUCCESS in 1h 33m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/453366c8ec554668b113aedd10e9e8b4 : SUCCESS in 36m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1dbc62792099418baef2ecdead24acd6 : SUCCESS in 33m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92c66c83fa854fd6b974570ce9287e65 : SUCCESS in 37m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a64c72ec7c868af20a65b5b1128e7dbc6ba756cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:47:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"09558e5668b5ec1cc1ecd5f06cb59c330cc1bc16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:49:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4784515df978471a80ee1b6a02e759d8 : SUCCESS in 1m 30s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93b35c5eda534f47813754f34e7f6eeb : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0720d67c6a579e3a6c71a5f3ae82938631420da","revisions":{"f0720d67c6a579e3a6c71a5f3ae82938631420da":{"kind":"REWORK","_number":1,"created":"2022-03-29 12:13:14.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/16/835616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/835616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/835616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/835616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/835616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/835616/1"}}},"commit":{"parents":[{"commit":"a16f516d68ad0ed1ccb521ef5b34bf290481c8da","subject":"Merge \"Fix ipv6 interface configuration\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a16f516d68ad0ed1ccb521ef5b34bf290481c8da"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-04 08:31:08.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-03-29 12:13:02.000000000","tz":120},"subject":"Pass timeout_dict to _get_haproxy_versions","message":"Pass timeout_dict to _get_haproxy_versions\n\nThe timeout_dict args was ignored when calling _get_haproxy_versions\n\n_get_haproxy_versions might be the first function (that uses the amphora\nAPI) that is called from rest_api_driver, it should respect the\ntimeout_dict passed by the tasks of the controller (for instance during\na failover a custom timeout_dict is used in order to optimize the\nfailover of missing amphorae)\n\nStory 2009761\nTask 44232\n\nChange-Id: I48a02780adf54b8d6d3bedada448a4f91e0539ed\n(cherry picked from commit 6d6a8bdf6c53c20ae5864ae61785ed902e468ae8)\n(cherry picked from commit 1da04d126a007416c0ba0822031818371266dd95)\n(cherry picked from commit 4e1cc209f7ba63ea9bf3eec1c0557da5bc2cac56)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0720d67c6a579e3a6c71a5f3ae82938631420da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0720d67c6a579e3a6c71a5f3ae82938631420da"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
