)]}'
{"id":"openstack%2Foctavia~835620","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I48a02780adf54b8d6d3bedada448a4f91e0539ed","project":"openstack/octavia","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-04-22 13:34:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I48a02780adf54b8d6d3bedada448a4f91e0539ed","subject":"Pass timeout_dict to _get_haproxy_versions","status":"MERGED","created":"2022-03-29 12:33:38.000000000","updated":"2022-04-22 13:36:22.000000000","submitted":"2022-04-22 13:34:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835620","meta_rev_id":"4ce60ac116cf8f4fe21ce21fca8b15cad5ff64f2","_number":835620,"virtual_id_number":835620,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-22 13:34:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-08 12:23:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-04-14 22:46:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-30 12:27:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-04-14 22:46:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-29 14:47:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-30 12:27:06.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-04-08 12:23:04.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-04-14 22:46:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"01b3f7d61f1179132ef5579ffe06548f50f3325f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-29 12:33:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86ff165e324ac109838239191d36a4a4b9fc3edd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 14:47:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a329d1eeb0d4d639fc250d9c59e1d75 : SUCCESS in 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aa711218db94b4e9183dd36f939d4d8 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d8688fa070c47b5b117d406b0bd2057 : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf354ff32ed34ce0b811d637ee2e4eb4 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78de2ad945dc45709f1a71a09ccfc776 : SUCCESS in 14m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8a74d4767584d63a21594f1b80652e9 : SUCCESS in 7m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2cb7c915e535493a8d1e43021ca1a802 : SUCCESS in 8m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f2b0445468b94eb88dc4b37d8b7430e2 : SUCCESS in 9m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e7ed9d93c32c4bc0869805e57fc86c1b : SUCCESS in 9m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/76069fb469d846b38391768d2a0dcd16 : SUCCESS in 1h 46m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/987f879ac92e4b6aa833a3efd5696be2 : SUCCESS in 1h 34m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/43a071a6d59f49f1893ed0ae9dd3e809 : SUCCESS in 54m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f0dda7e7b87a47db8158fabdd01d7e9a : SUCCESS in 1h 03m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d43d82f4092048a9b3451090f95f7124 : SUCCESS in 1h 16m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e9b29e4e18564d4281ce1bbd03abbde6 : SUCCESS in 1h 20m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4028dae0527843d5b1f301015deddfc5 : SUCCESS in 1h 45m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2bb4712ba93444f79f56b21cccb2f4da : RETRY_LIMIT in 7m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/38da97afc9fe491e8de65096b95482af : FAILURE in 2h 06m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f2351b74ce8634e04d4dd4302af11f3b1e6dd5b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-30 12:27:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"190d9e672b981646719a0af4f96e62be132862f0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-08 12:23:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e1d94e1889675fe1268732b1cd0abe9b4c736f69","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-14 22:46:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5e327debe0b2ff9f40481a8f287d92524311b73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 22:47:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"229aef434d35875feffc750178f85dad93d1c126","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 00:26:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd1ff65f3f6b422ba44119c96b91008f : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c13b22c45e7a4cbca62ee7406d4d553f : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0f01aeae01c4e4b8566c97c447d6e0f : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/718ebb12131d4b9db192da811b572cc6 : CANCELED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c1ce347aa4d48f5a301d4267a9ca901 : CANCELED\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eadc581f382f463aaf51ebe58d2b1cb6 : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eaecd66efc88417a88ea44ba74ed7736 : FAILURE in 9m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7da3b51c8ed3405d9b3a3afa68aad4af : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/994e849f0472408d94faf1b452b46155 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/48146670f7fa4a3ab0009ee4b215171b : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd04df00b826467ca2983d56526e72ff : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e83917752429419ab0546cfd6cb0506e : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"7b719437ba34cbc47918e58163a4a66550c1d68d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-22 09:57:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"272bb084e3d51ebbed8927d67dcc52ccf2dc3de3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 11:57:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/738083eb844f4afb965b0a34287a1406 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cd57eb667664c1088d602570dcb2add : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac49453f568c44c2adc4430319433a0b : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4a603d93e2e491b9cc7e926bf85a47b : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4355ea9177d04aba8426a413d93dd406 : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09cede4d414e42f5bcd50556fc8401fe : SUCCESS in 5m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fcd4967e046640039eb34b2037312029 : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b6e2b42210ac44dca6462f81ede684f6 : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ae95f5b59af4f96be5c93f1152d6e48 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/171a85765c6d4bb68fd4775a47e5598d : SUCCESS in 1h 06m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5cb749c37f294e66b43fe0e7606839ca : SUCCESS in 1h 55m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f60376e338c44e99ab5009de927dfdd1 : SUCCESS in 35m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3cdb9e13408c421e9ff098f74fea3000 : SUCCESS in 33m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12598ab3958c43a0a737eced33bdb426 : SUCCESS in 32m 34s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0f686a1f3c8a46deab3096fa089acea2 : SUCCESS in 1h 35m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/46530b1591e342ed9b14cfcbeb92f511 : RETRY_LIMIT in 7m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c593129ba89a41e681f7b1da036d1b4f : SUCCESS in 1h 38m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f660630d5443450cc24de2caf099e0c2475ed62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 11:58:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd7e893bd20c28d2310fb683582463fdf9c79d4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 13:34:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7e12c279b4b43cd9651009f69d89556 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56431e9459464d79b9bc01d8a3d2ab36 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce92b308d0b145958559256e2edb4252 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/843561510a8e497cb4686fb7f3aa1149 : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7db81c9560d94171bfcac8c307632e26 : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b0b3841beb3f4830842915001f824410 : SUCCESS in 8m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ed6b3351c7b141738e5d5cafef37d63b : SUCCESS in 1h 00m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ae38106881848bc92b2e2c7342bb4e2 : SUCCESS in 1h 36m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1af5706b442d431cb2cdc97bf409f526 : SUCCESS in 36m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/843fce964a6e499fa51e4d29f19557b6 : SUCCESS in 30m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3cccd5a7db7b45049dcde32cdecc5800 : SUCCESS in 32m 10s","accounts_in_message":[],"_revision_number":1},{"id":"7bbb8603c17ca9af21a1b4dccb91e369ea783d16","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 13:34:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4ce60ac116cf8f4fe21ce21fca8b15cad5ff64f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 13:36:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a43fbe56ec764398a521d215f1ce647e : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8ac8290db4845518a2157288ff7d921 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"94dea3a4e8bdabf02192e5cd2155ac3935852e91","revisions":{"94dea3a4e8bdabf02192e5cd2155ac3935852e91":{"kind":"REWORK","_number":1,"created":"2022-03-29 12:33:38.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/20/835620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/835620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/835620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/835620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/835620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/835620/1"}}},"commit":{"parents":[{"commit":"3d56e7ad20ec3bf3280ca10f7e3d3754ea197b71","subject":"Merge \"Fix unplugging member ports\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3d56e7ad20ec3bf3280ca10f7e3d3754ea197b71"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-04 08:31:08.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-03-29 12:19:52.000000000","tz":120},"subject":"Pass timeout_dict to _get_haproxy_versions","message":"Pass timeout_dict to _get_haproxy_versions\n\nThe timeout_dict args was ignored when calling _get_haproxy_versions\n\n_get_haproxy_versions might be the first function (that uses the amphora\nAPI) that is called from rest_api_driver, it should respect the\ntimeout_dict passed by the tasks of the controller (for instance during\na failover a custom timeout_dict is used in order to optimize the\nfailover of missing amphorae)\n\nStory 2009761\nTask 44232\n\nChange-Id: I48a02780adf54b8d6d3bedada448a4f91e0539ed\n(cherry picked from commit 6d6a8bdf6c53c20ae5864ae61785ed902e468ae8)\n(cherry picked from commit 1da04d126a007416c0ba0822031818371266dd95)\n(cherry picked from commit 4e1cc209f7ba63ea9bf3eec1c0557da5bc2cac56)\n(cherry picked from commit f0720d67c6a579e3a6c71a5f3ae82938631420da)\n(cherry picked from commit bab9a45325bdce43bd20d272130034a220868e47)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/94dea3a4e8bdabf02192e5cd2155ac3935852e91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/94dea3a4e8bdabf02192e5cd2155ac3935852e91"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
