)]}'
{"id":"openstack%2Foctavia~835779","triplet_id":"openstack%2Foctavia~stable%2Fxena~I658a4097e120d4f6a49fa1a7945a8d999709d1bd","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-03-05 19:15:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I658a4097e120d4f6a49fa1a7945a8d999709d1bd","subject":"Switch centos8 jobs to centos-8-stream","status":"ABANDONED","created":"2022-03-30 06:38:27.000000000","updated":"2024-03-05 19:15:22.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fad586e1650fb925c69562cb2d5a02eb64bf7042","_number":835779,"virtual_id_number":835779,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-30 09:02:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":-1,"date":"2022-03-30 09:14:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-30 09:02:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-05 19:15:22.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"065b0306d5fb8288b50c42c84b336d64fe8068bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-30 06:38:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56e98225241161696add0c40b24bb9e076fe3132","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 09:02:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdd450ee34104fc88fbb3fffeb384f09 : SUCCESS in 11m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc86fec722124bc98b4934a4ebe106f3 : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc0e3cdeba494b499d1ea0c95ca7af8f : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbe8a5169d1c47c2be2c6e3c1b68f796 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91568263f48a4cd4a02f4e498ca84fb8 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9e6274ac7424a33bea017aef757f4a8 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24285f9dbcc84d06a6f6382fc881eca8 : SUCCESS in 13m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3d3ab46112364ab4915d11a66ab5ecad : SUCCESS in 7m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f338a1c5f2374cdba24702d40a023c03 : SUCCESS in 10m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9555044208d240e69f40e50ee572b5d7 : SUCCESS in 7m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b4881d71736746f094f0d4028d1cc6d9 : SUCCESS in 9m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d8d92034ec642acb6124c6bc791f858 : SUCCESS in 2h 18m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a7bc6d6f94244b5a9a925fab5b271f8 : SUCCESS in 1h 35m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/613e999aebb844fc9e0cc4a335468ceb : SUCCESS in 39m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e84bcf4c5e6e4aa8a65bcaba0551df40 : SUCCESS in 57m 06s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/a16b16514f334cbc807ff2ed7c112029 : SUCCESS in 1h 45m 40s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/cf26248fd03f4499a081c5041cc7354d : SUCCESS in 47m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f6de822a7f904925b0ad877dd1f9b15d : SUCCESS in 33m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b2f3dff60c5d420198b11ee62cd1f7f4 : SUCCESS in 1h 40m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a6c0f7de917049b489e487edd63c8257 : SUCCESS in 2h 05m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dd08b11709ff44f2882082444dc01fa2 : SUCCESS in 2h 09m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eb9269a3e4b34cba888dafd0bd1481a0 : SUCCESS in 1h 31m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ac852ee4786f4d1b8d76969edb406dee : SUCCESS in 1h 25m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/51807124b2d74043903184d130b8e4ba : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a405ef8732e42fdff2db3aea57b3990f723c7be","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-30 09:14:08.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fad586e1650fb925c69562cb2d5a02eb64bf7042","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-05 19:15:22.000000000","message":"Abandoned\n\nstable/xena branch of openstack/octavia is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/xena if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1755832a309c84050499d66768060f0ff237e29c","revisions":{"1755832a309c84050499d66768060f0ff237e29c":{"kind":"REWORK","_number":1,"created":"2022-03-30 06:38:27.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/79/835779/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/835779/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/835779/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/835779/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/835779/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/835779/1"}}},"commit":{"parents":[{"commit":"f9af20e80a2780dc3f1de06786ccfce52f8ce8f9","subject":"Merge \"Fix ipv6 interface configuration\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9af20e80a2780dc3f1de06786ccfce52f8ce8f9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-13 05:52:53.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-03-30 06:38:09.000000000","tz":120},"subject":"Switch centos8 jobs to centos-8-stream","message":"Switch centos8 jobs to centos-8-stream\n\nChange-Id: I658a4097e120d4f6a49fa1a7945a8d999709d1bd\n(cherry picked from commit 76617f883e535a77008aac80d5350d0f0c889f8d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1755832a309c84050499d66768060f0ff237e29c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1755832a309c84050499d66768060f0ff237e29c"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
