)]}'
{"id":"openstack%2Foctavia~836314","triplet_id":"openstack%2Foctavia~master~Idd6480bc0a7de7c42b34601172c2dae50077aa48","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-08-18 15:18:18.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-04-27 15:48:20.000000000","reason":"Tom Weininger replied on the change"}},"hashtags":[],"change_id":"Idd6480bc0a7de7c42b34601172c2dae50077aa48","subject":"Fix unclear error messages in the API","status":"MERGED","created":"2022-04-04 11:30:03.000000000","updated":"2022-08-18 15:19:28.000000000","submitted":"2022-08-18 15:18:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836314","meta_rev_id":"4b84cfe172aca6359b028ea427ce047289be0a07","_number":836314,"virtual_id_number":836314,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-18 15:18:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2022-08-18 12:41:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-08-18 10:05:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2022-08-18 09:45:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-18 09:32:27.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2022-08-18 12:41:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-04 13:42:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-05 13:43:40.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-06-28 13:56:09.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-08-18 09:45:55.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2022-08-18 10:05:53.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-08-18 12:41:11.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"33222bbbe53eadeb2cbe155f10781c26a6076617","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-04 11:30:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed8fadc8baed3545de1b889fcb05d81f6b82537d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 13:42:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10121434a7374b47ae96c8a0e5cac541 : SUCCESS in 10m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc7b06455d2b4629b19f5c061798fee4 : SUCCESS in 12m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eacd640a40204047a155b6332057695f : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9554abc132954c0393e439ffda1f3b8e : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a08cb34454f4a5f9993dfffd2e015cb : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be5373b828984e6687e503cf5534e513 : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/174d3a309ab9452f82587cdfc2d409b6 : SUCCESS in 11m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27e45ea0684041ec94eca90ffd5f66a0 : SUCCESS in 5m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e6fd2245b40a4bf698ce425cd7350478 : SUCCESS in 7m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b16ac472c1274e8e90f4ed592205ea05 : SUCCESS in 9m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ae42b2eddc0d406a84c0209934aad8ba : SUCCESS in 6m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e0c4557992247d3a72d0d970a4af4e4 : SUCCESS in 9m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/28e91008d9e94e66923c308503a993e3 : SUCCESS in 2h 06m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/718432af5cfa4f788fb842a30c4aa8ae : SUCCESS in 1h 19m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5498591810f64ed287b453a6b382a7de : SUCCESS in 38m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4e8e4a79f9f342908daabaa777918beb : SUCCESS in 52m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8954407194ed4488a71f25b922098c02 : SUCCESS in 49m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0022eab6db604da8927a0ed0db6209b6 : SUCCESS in 33m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5c4dca4de43446f2b179a753b978236c : SUCCESS in 1h 43m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f58adba9ea284459a78b537c051e2968 : SUCCESS in 2h 04m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/993e3a86881946e18decd23f6cb5dae8 : SUCCESS in 1h 36m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d36248df4f6046419aa42f59ec8ea8dc : SUCCESS in 1h 23m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/744e4af456bd42ff90464384d5fb0a7a : SUCCESS in 1h 29m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/518f138cf3804bfb90e94ef61b0c4dd4 : SUCCESS in 1h 49m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6aa1123d6624f3fe9c6b748c286b76eefb8c08e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-05 13:43:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b64fb518c343f2638517b98d7f0471424862f5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 13:07:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fc6e6b910a65a3b7ba1922105960bbc99ac403d1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 13:07:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ff96724ba42a80cdaab541040d3b92ce3b36fd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:38:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23e60233efc7430fa593b376808772ae : SUCCESS in 12m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0eec1f2ea974a329be58d5b14174578 : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a81b7758ceef47b893dd8012a671c01d : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e54b501283e043de8dd73d191ccf63b7 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e956a049e81d474b8df79b74b225e8b2 : SUCCESS in 6m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77faba4b150c4a888b901504de664513 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df9722552b4241a793d6b340f3cc754b : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fce6ddc6a1a944849d817afb9c4cd28e : SUCCESS in 5m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c8df9f0d6f5944fa9eb6590d99fe11be : SUCCESS in 7m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0cdf2d241b4242e0ae0c86a64e4459d1 : SUCCESS in 9m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/405894c5b69e441fa7e2728676c5945b : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ff66f309ad4f4f88b656dbcfddcc3098 : SUCCESS in 9m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2507e65e19f545c1ad5f198ac7def569 : SUCCESS in 2h 27m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5c23784b9964d039ef52e08242cc987 : SUCCESS in 1h 28m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/062533343fe04253a893423cdeaee603 : SUCCESS in 41m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/73c8135a06f845a8b4b0dfe30ff627a0 : SUCCESS in 1h 12m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3fa2215cf38d4f18a31e5fcfea9e0b93 : POST_FAILURE in 17m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1e8d191c002400f83bed2b516862e15 : SUCCESS in 35m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7c219113603d4234b548485a94501c4c : SUCCESS in 1h 50m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e46509b244a1460bb8fe2c2e47df5008 : POST_FAILURE in 26m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c90ca53e50a149959eee7f2e99b84840 : SUCCESS in 1h 43m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e6ebfa1c8bfe43488999e9b82bae0943 : SUCCESS in 1h 23m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/04fb242b32824c31a18c578b84f966cb : SUCCESS in 1h 31m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/92864c94b6474624b5bcb86e97e6a4b7 : POST_FAILURE in 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"06dda6b7a6f45d3e333da316e143d08f5f468764","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-27 15:48:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"deff523767a64a57cebd0b6ef7a462a4defdb195","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-06-28 13:56:09.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a7a1470fb7915b411acde8438d7f162df6841e84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-17 11:49:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c9ae57c0313defaaeb5fd47f4ae79b8f5dfc0b61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 14:19:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76772ab6eac4428e914de58d08f9c61c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cfa308b7d364abeb151591911bbd83e : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/113aa2ef6c3e464da3a165c68a0140e7 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8dfae5375894d16b2d8d8d48736888d : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96f2345f65de491da70b7f4b830cdd4b : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01376899204940b982c235797736d084 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ee40da0e8fe41d080e4537cb5c2111d : SUCCESS in 12m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01d13af57aca4a2085505a07272f2392 : SUCCESS in 5m 08s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/62089dbe45cf4ac7bb3778f0080325a4 : SUCCESS in 5m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fa8627b0555541df99200f4bb8980227 : SUCCESS in 9m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e5ad1205354346feb7cb86328f7a4c77 : SUCCESS in 5m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7ffe741e18ed4da59ae02447de2ff55a : SUCCESS in 7m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/992ccb67bb1e4c63b55aa70d79bc6200 : SUCCESS in 8m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/88e11f17c4e84f9882da5ca91bec5f77 : SUCCESS in 8m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5acd7c2c42ec4697aa42c9222521b3e3 : SUCCESS in 2h 24m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc208b270a53400d90e3de21e4973116 : SUCCESS in 1h 47m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b09e25f0a83e401f81c8bb04fc87ed0e : SUCCESS in 43m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5e3c46568d4442329fe4bbd997d1f44f : SUCCESS in 1h 07m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3b209ff310124126bef798911b91f19c : SUCCESS in 1h 28m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/865bb6014ab4439d97bfc66d00d60734 : SUCCESS in 42m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2020b83ec3a8441ba54695cc6779568c : SUCCESS in 1h 34m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4689ed2c8df74daca13da9cd77c6dbbb : POST_FAILURE in 1h 10m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/05a12d5f4edb40c0bc8240e5379a91c1 : SUCCESS in 1h 43m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/59c728e63b3049429926047f64db33d7 : SUCCESS in 1h 47m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8d841f92a5754eadb8a88f3b61b805bc : SUCCESS in 1h 28m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/62d3b3a93e734f60958f392ae570ea1e : SUCCESS in 2h 19m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"16981a3fc0af91f30f9afe9f5d3bf7ec78e77da1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-18 09:32:27.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6c2745187cbfc872690784818f92b2b61baa5847","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2022-08-18 09:45:55.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e3b90b663182bae598b2dc8472bd1e37120f4965","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-08-18 10:05:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5326c11fe8b7928c61fa7302162c54fc50f4bbf2","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-08-18 12:41:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"31470f90bfb962eb242ea5a6f0d9c66516f2a28e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 12:41:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8d2a7e36864208f2b5bdea42c7f10de8c42f70c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 15:18:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c0881186704468fa8f5d36a705a9152\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcd35e27ccb6432aa9b22605ade7c7ae : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa5269769f35497096415cdeeeaa76ae : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a01b76a8223a44eb82ebd3998c34e736 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a46cdacafb0e480e8b0c82e30acb4dfa : SUCCESS in 12m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75de8ea2707e42dca41d145dc3513ec0 : SUCCESS in 5m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c9365af3b88e42fc8d7319e8d6312082 : SUCCESS in 5m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ca046625f8cc44a49b1c696c5eb3af79 : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2e1968af988841e88c26cd25af5095e7 : SUCCESS in 2h 31m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0f33c57c69f43a6a7136f1c875952f6 : SUCCESS in 1h 49m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/feed6ae48d624483b6664dab460906cd : SUCCESS in 50m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/82ee5a26472a4f2ba345469a2100003a : SUCCESS in 1h 27m 58s","accounts_in_message":[],"_revision_number":3},{"id":"a42fe3436a1fe428d79420476b71f9a21288b519","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 15:18:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4b84cfe172aca6359b028ea427ce047289be0a07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 15:19:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/110e04ba220549218956e9562c697d19\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6476c79e84944d29334440a753e43c7 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d93b46b5b89b4ecaab8b4680f70f15c0 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"57e006eac321b210c98917019572b3576aefd97e","revisions":{"92b48926ea8222173ff75928cc1ae4449b3d6552":{"kind":"REWORK","_number":1,"created":"2022-04-04 11:30:03.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/14/836314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/836314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/836314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/836314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/836314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/836314/1"}}},"commit":{"parents":[{"commit":"309c7e61fe1d9f15c1d1fcb6665af610ade94dc7","subject":"Merge \"Pass timeout_dict to _get_haproxy_versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/309c7e61fe1d9f15c1d1fcb6665af610ade94dc7"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-04 09:20:20.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-04 11:29:30.000000000","tz":120},"subject":"Fix unclear error messages in the API","message":"Fix unclear error messages in the API\n\nSome API calls may have returned useless error messages such as\n\n{\"faultcode\": \"Client\", \"faultstring\": \" is not a valid option for \",\n \"debuginfo\": null}\n\nwhen a DB constraint failed.\n\nThis commit improves the error message by reading the name of the\nforeign key that failed from the exception.\n\nThe value of the parameters should have already been validated by WSME,\nthis exception is only thrown when there is an issue with the database\n(corrupted DB, empty table or missing fields in a table).\n\nStory 2009957\nTask 44919\n\nChange-Id: Idd6480bc0a7de7c42b34601172c2dae50077aa48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92b48926ea8222173ff75928cc1ae4449b3d6552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92b48926ea8222173ff75928cc1ae4449b3d6552"}]},"branch":"refs/heads/master"},"61af03320472d99bee91feacf81d7b96a0d7e8cd":{"kind":"REWORK","_number":2,"created":"2022-04-27 13:07:06.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/14/836314/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/836314/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/836314/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/836314/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/836314/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/836314/2"}}},"commit":{"parents":[{"commit":"309c7e61fe1d9f15c1d1fcb6665af610ade94dc7","subject":"Merge \"Pass timeout_dict to _get_haproxy_versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/309c7e61fe1d9f15c1d1fcb6665af610ade94dc7"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-04 09:20:20.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-27 13:07:00.000000000","tz":120},"subject":"Fix unclear error messages in the API","message":"Fix unclear error messages in the API\n\nSome API calls may have returned useless error messages such as\n\n{\"faultcode\": \"Client\", \"faultstring\": \" is not a valid option for \",\n \"debuginfo\": null}\n\nwhen a DB constraint failed.\n\nThis commit improves the error message by reading the name of the\nforeign key that failed from the exception.\n\nThe value of the parameters should have already been validated by WSME,\nthis exception is only thrown when there is an issue with the database\n(corrupted DB, empty table or missing fields in a table).\n\nStory 2009957\nTask 44919\n\nChange-Id: Idd6480bc0a7de7c42b34601172c2dae50077aa48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61af03320472d99bee91feacf81d7b96a0d7e8cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61af03320472d99bee91feacf81d7b96a0d7e8cd"}]},"branch":"refs/heads/master"},"57e006eac321b210c98917019572b3576aefd97e":{"kind":"REWORK","_number":3,"created":"2022-08-17 11:49:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/14/836314/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/836314/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/836314/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/836314/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/836314/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/836314/3"}}},"commit":{"parents":[{"commit":"f64f2cdc6eb0732f559323b89ec3c448af316a0f","subject":"Merge \"Workaround for functional tests timeouts in DriverAgentTest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f64f2cdc6eb0732f559323b89ec3c448af316a0f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-04 09:20:20.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-17 09:47:13.000000000","tz":120},"subject":"Fix unclear error messages in the API","message":"Fix unclear error messages in the API\n\nSome API calls may have returned useless error messages such as\n\n{\"faultcode\": \"Client\", \"faultstring\": \" is not a valid option for \",\n \"debuginfo\": null}\n\nwhen a DB constraint failed.\n\nThis commit improves the error message by reading the name of the\nforeign key that failed from the exception.\n\nThe value of the parameters should have already been validated by WSME,\nthis exception is only thrown when there is an issue with the database\n(corrupted DB, empty table or missing fields in a table).\n\nStory 2009957\nTask 44919\n\nChange-Id: Idd6480bc0a7de7c42b34601172c2dae50077aa48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57e006eac321b210c98917019572b3576aefd97e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57e006eac321b210c98917019572b3576aefd97e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
