)]}'
{"id":"openstack%2Foctavia~837547","triplet_id":"openstack%2Foctavia~stable%2Fxena~I550c7d2563a2d3ada07e2162383f59f104bcbbd5","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-04-21 06:43:50.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"I550c7d2563a2d3ada07e2162383f59f104bcbbd5","subject":"Restart rsyslog from cloud-init","status":"MERGED","created":"2022-04-12 10:49:57.000000000","updated":"2022-04-21 11:12:14.000000000","submitted":"2022-04-21 11:10:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837547","meta_rev_id":"461883b7b412071645c4ed68d7cb3cf43dbb6413","_number":837547,"virtual_id_number":837547,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-21 11:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-15 11:44:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-04-14 22:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-04-15 11:44:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-12 13:27:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-14 22:51:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-04-15 11:44:30.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"3177476491286af3f2437070e54eee4c924a89e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-12 10:49:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89b639d989812aba52c611dc3d22c816461ec322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 13:27:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e41b12c78027436197d34f6ee5358487 : SUCCESS in 10m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9117446c1e9c4f369dc27b5a80812fc3 : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f331afea7a4a4675a4af7baf02439ab9 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a569a070c5f146289c4efe6c88499df7 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97a473c13f9d492a80d40320fb04e038 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b2072341e1b46809dd1ede956d5886e : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/069b3456313d47da9722dc14da926b22 : SUCCESS in 13m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/951ebe23597f4676b2cc9a4db51879e5 : SUCCESS in 5m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e8164cc2a7774341978a47bc37a258d2 : SUCCESS in 7m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a9de4a5a10994bf1bb4905a81cf4fb2a : SUCCESS in 9m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dbde0cfb2d0f48f09fcd36e29a208b2f : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e002ace6f26a45bfb4386a03538bfe2e : SUCCESS in 10m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7c52d4b11b784eb587da466fad220c48 : SUCCESS in 2h 26m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54b5d42dedaf4852918d46d03b8c8ecf : SUCCESS in 1h 19m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3bf73cc698654de993668fb45d56ec93 : SUCCESS in 35m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b540424741fd4bda971a786e083836cc : SUCCESS in 48m 53s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/1ce770529fa2486b83d4217e86e7c85f : SUCCESS in 2h 22m 51s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fab62660d53f44a4a641d388529b2969 : SUCCESS in 49m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3106a2f7fd4647639444bf431dc052e0 : SUCCESS in 35m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c8871aff89774ab3955f0b3ade84c765 : SUCCESS in 1h 30m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2d3485a98e994925911237cd9972e19f : SUCCESS in 2h 05m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1832b17b891940c7bc61c92a47282c39 : SUCCESS in 1h 31m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f0120588fd544b0893f6f4125f5aed1b : SUCCESS in 1h 26m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6f04a4f0a9054004a3c2979a31b7533a : SUCCESS in 1h 36m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b099593c242b451ca3edc1da4692afc9 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84dd6cc10689e0409ec71b2bb69274c9e0288f48","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-14 22:51:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f5de1c1fe4ae27ad91c35ddf179f51a54e56d62","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-15 11:44:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d80057ff1ead910e863233b533fcfa87c78d452e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 11:45:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57aef652f3e93afc8ce3655095ad555721eb7c8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 12:50:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/408f4878d8984a6ab20c7b45f8e79d53 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84c827ea9e0c479f805a44905111454d : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/815c8472225d4caaa229bab99ae1802c : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40f98b70dd264b4d9f10111babbf0ec0 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2add2418e70040ff984068d14cfc5530 : SUCCESS in 12m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aaf355871e234e2bae4a287209455982 : SUCCESS in 4m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e7bab1b0af7944758bc02f649315abfe : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d48f96c619b413aa130cfeb70ed1d23 : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/25e68287938647528c122dffaf565b03 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7580f29cde014c5badcfaae12c8c76a5 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ad44895afe264d528bb249984d2c8acb : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b834e767cdf44a2d96eb91c1f78b271a : FAILURE in 24m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6ba4b633ee5dea89447a57cea400c101e310dcd7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-19 12:31:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6a7d4066d27e65873e08eb517a2c223b88d811f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 15:22:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56901d8154234a21aabb0390b1f99f90 : SUCCESS in 14m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/864a978a846c4188a1dae2e081bc5b17 : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0a4df40a3f3445f83f82b23a08427e1 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11b57d7b20c24e568b0eccf4821bf95e : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e498c9359c7d4ae0bfbf2b426d9fdb8b : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03d9687fa5a443b1ad03f79d6b8b2a2f : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5014e4a65ed2447f8e5e85b4c4e15903 : SUCCESS in 13m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca4e40327a884ccc9127bcb29c160aed : SUCCESS in 5m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0ca33015f82f41b4ab3b0cd60285f2fb : SUCCESS in 6m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c089cee0aba742b594f65058b3fedafe : SUCCESS in 10m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e5f93ce44b04499cb63439c4115d9b95 : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/defd3346d7574055b00b524b02309384 : SUCCESS in 8m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bfc2e060aafa455c9db2852573add2c0 : SUCCESS in 2h 04m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4cc8a86a1ad242588ae8807ae1e11cbe : SUCCESS in 1h 31m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bfe6b2b32db3417db8eb437d77342232 : SUCCESS in 35m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/83c5a533ca5245d0b373f5790415af9b : FAILURE in 24m 57s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/501295145be64cc49b2a623000302c6b : SUCCESS in 1h 58m 52s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2bae15c7d6be4824903316d557e24ad5 : SUCCESS in 51m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8be20c0b521f4893857ffe518f49c3a0 : SUCCESS in 36m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/eb28cc97cced46cb979433e97bab605b : SUCCESS in 1h 38m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/21a2501f7a684c02ad8a7be3bd1a2301 : SUCCESS in 2h 18m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a4000a06c9074db99537eda72b354724 : SUCCESS in 1h 32m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/140714bc1f8c4a488970451058304be2 : SUCCESS in 1h 41m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9a7017cfe3fd4e40a8062d4de77be1d1 : SUCCESS in 1h 21m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f41fd32842204a88860bf1a4fee9da28 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc699f4c4451678cd5ba92dcfaf73baba65bd343","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-21 06:43:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"701fd4c14a305ccb7ae80b232e911acfadb2807b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 08:55:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba52336ce74d4b3f84dae61c0cbf1ec3 : SUCCESS in 11m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfb94db2412a40b195229cec7de1597c : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4c08d3a9c2242af9c0a262d79c9c80b : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/088e8e95db6b43fdb649019ddf0d718b : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c15e16c284149af9c9d9e3c17bc81d7 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aeff78c0951a45c0b09c98b207423407 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fb526136e8f443ca7655256c3087018 : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2aaa5097408a4fdda58a0145c8d3ff94 : SUCCESS in 5m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/53cc89dc989a4bc786a0f15830eed410 : SUCCESS in 5m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/502e460881b242dc959f1bf883827b50 : SUCCESS in 8m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f2c38e749520448e99fc97a7e2a83d95 : SUCCESS in 5m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b84c4cca3c544aa984397550ba607ea : SUCCESS in 8m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e402c1710c444ab7b7b5ac86cd94178a : SUCCESS in 2h 09m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3a518a5d968947149266702fba26c337 : SUCCESS in 1h 51m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0a5a8da4ff9a4ef3ab06051a962a6c2e : SUCCESS in 35m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/75fccd5c4d0d4061a90b3a7596e79a35 : SUCCESS in 53m 52s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/94b8f7269a1f4ba6886cd0b1b09d530c : SUCCESS in 1h 51m 34s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b4e5fa44eec9410db104e142210b871e : SUCCESS in 43m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0eec8698dfab42ca9e74e1ba2d83bbf8 : SUCCESS in 34m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b8536c340e144ba1bfb2b7b07500f8ba : SUCCESS in 1h 59m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff0e2aae0f834a7c8ab58c0b67ca9811 : SUCCESS in 1h 58m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2863246b44464c059107a2aaa8dbb9ac : SUCCESS in 1h 57m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8ca21a4c3c404fa6ac9b7dcb1ca00431 : SUCCESS in 1h 57m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d6f10ebd6fb04a8cbc04923931b6c4b9 : SUCCESS in 1h 47m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cd724bfb6adb4af0b82960d8a28b63d4 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f2b4a8274b722cdb77558322dfd77ab4c77d0a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 08:56:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"11f3a3e1522151a8072ad311bbc05a0e592dab66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 11:10:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd6a3dbf42864232a90c5cecb0a42f4b : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fe39340e9144d00841649185553bd13 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88e9b021909a4959987f9a3fb1231478 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f841b7595f35451c84278e838c4a7426 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6828c3c5ab94d5d9e790c8d989d876f : SUCCESS in 13m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b1926cdd4384fe1ad27450e47be8e46 : SUCCESS in 6m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/75413e6d09744edaa4214a1f3bd73f71 : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/17856c384f25437e920b8747ccfd9d01 : SUCCESS in 8m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2ab004edad1f4d588dc2a1cf47385a86 : SUCCESS in 2h 12m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/efc8ccc59c7842d89a9b23fed7a6623c : SUCCESS in 1h 35m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/01eb7af63e4b40b899e53af887594a1a : SUCCESS in 35m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7ae67c0c898e4de0ae1e2b6f640d1504 : SUCCESS in 49m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7712f46a5dc1fb435e7f07afd7f0e35b4450abda","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 11:10:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"461883b7b412071645c4ed68d7cb3cf43dbb6413","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 11:12:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49472944d10743a3a4c29dfa7a46eaab : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b17fa423aca545bf9630c18e99c82d2a : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"220f13f537cba50f0fb575b0e3b5c39c85e8a0b4","revisions":{"220f13f537cba50f0fb575b0e3b5c39c85e8a0b4":{"kind":"REWORK","_number":1,"created":"2022-04-12 10:49:57.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/47/837547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/837547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/837547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/837547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/837547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/837547/1"}}},"commit":{"parents":[{"commit":"f9af20e80a2780dc3f1de06786ccfce52f8ce8f9","subject":"Merge \"Fix ipv6 interface configuration\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9af20e80a2780dc3f1de06786ccfce52f8ce8f9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-27 14:19:00.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-12 10:49:27.000000000","tz":120},"subject":"Restart rsyslog from cloud-init","message":"Restart rsyslog from cloud-init\n\nOctavia uploads a rsyslog configuration file using cloud-init but\ndoesn\u0027t ensure that rsyslog is restarted. In some cases (CentOS 9\nStream), rsyslog is started before cloud-init and doesn\u0027t use the\nprovided configuration file.\nThis commit ensures that rsyslog is restarted by cloud-init.\n\nChange-Id: I550c7d2563a2d3ada07e2162383f59f104bcbbd5\n(cherry picked from commit cbffae6eb684629a19844b5f2f4b7523b77a725e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/220f13f537cba50f0fb575b0e3b5c39c85e8a0b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/220f13f537cba50f0fb575b0e3b5c39c85e8a0b4"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
