)]}'
{"id":"openstack%2Foctavia~837548","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I550c7d2563a2d3ada07e2162383f59f104bcbbd5","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-04-27 12:58:24.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"I550c7d2563a2d3ada07e2162383f59f104bcbbd5","subject":"Restart rsyslog from cloud-init","status":"MERGED","created":"2022-04-12 10:50:14.000000000","updated":"2022-04-27 17:16:03.000000000","submitted":"2022-04-27 17:13:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837548","meta_rev_id":"d29b51a1aafea3763f4bcb52dc1d923843f18731","_number":837548,"virtual_id_number":837548,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-27 17:13:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-15 11:44:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-04-14 22:51:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-12 13:19:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-04-15 11:44:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-12 13:00:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-12 13:19:00.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-04-14 22:51:03.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-04-15 11:44:39.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"8bffa8470fd1e8e16072ef1e0472e6b7a36abf77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-12 10:50:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"038b10dcca64e56a039df4352dcdcd25fe2423d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 13:00:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b264e244269f4d11ba0d788d0a6519d9 : SUCCESS in 9m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8d6a2af4ecc46e8826815e73d17f696 : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab4f33247ec945deaf731819daab1368 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dde7e36626a24071a900ef6954772cd4 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55dceeda208c4bdf836f489074f1df0c : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bf372df99944f4985af99342822cc08 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb7c298e0445443b8a5299e2aebc390b : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/327a77e445df444a80e4d3e1f3e01d69 : SUCCESS in 6m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7d6ceea0c24a48c983d0a1597c51e7d9 : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1354d8e3fb5b44d0b4ac70fcec48160a : SUCCESS in 10m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6dc88a52b1a04498b2afad977231ce8f : SUCCESS in 6m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ec032044e044c16b1166ce863286e5a : SUCCESS in 8m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3a96eeeeba7b45b4a3e2bad27ea2ac33 : SUCCESS in 1h 38m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20737da4129a45a8a707bfcadd42b2d7 : SUCCESS in 1h 33m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/490e671e92504d85af678d57225407d9 : SUCCESS in 36m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d00194825db54be3acb54c426902d09d : SUCCESS in 32m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b8bcd48f83774a7fbfe3f11ec0502bc6 : SUCCESS in 34m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a1df1fce35a849879240fa12bc705ed7 : SUCCESS in 47m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c2e1b5b2d51f4e8d80409223bd9c3d90 : SUCCESS in 1h 29m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e3e2db1177744086bbbd304aa87c81a3 : FAILURE in 32m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6f8178a3ea62455cb2dae2e97af96798 : SUCCESS in 1h 39m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/69a65f1127754f359071800a32fcd3d5 : SUCCESS in 1h 33m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e13b9adaa20d43ed98fb3e3ccfb0957c : SUCCESS in 1h 35m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f44022d3d6f54ef8a47f147b9c5aeb58 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52cdcb1c37581aee52b7488751f7417abc651534","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-12 13:19:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ed27e8d1de2cb1be88b2aad44c80f195d3ecf7cc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-14 22:51:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"377e732e062d5a384ca91813cfe846a3d627d1e5","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-15 11:44:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ed6bb7e7e583c68e011166873d1c58334a1f0ca4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 11:45:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4cf8a94a3cfd58abe1f4d6aa385d43f13af31b61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 13:04:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad1bcbad01d04cb4b62af118f7e23050 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34373aaf352c4438b8d4770248fb3921 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76971be1abcc491387dd8abe0eacb1b4 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b86e374576204da6b6e98621aca72c40 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0cc6f4d6c4a45bf979c5170c897b3bd : CANCELED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e36b0c1839f244a5b834f16c2e8ade73 : SUCCESS in 4m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0ebce8bdd5b447d4bd48a3f00fe25429 : SUCCESS in 5m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/27354b01bd164dcc8d83480c603fe1a5 : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/22cdbda549724107b35cede816c58277 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9cdb23350b0d475f8aafb553e8fe30b1 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b20386448f2c44afa85c8c527f34d2b9 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5d9d2ce5a0404abd80931e8f5b300afe : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afb4737b44d5426c95c07a50522e89c6 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/03a6129f878c4e82a3b095166ed0b0e2 : FAILURE in 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"eec26618f27d20a43006d41c071f5dfedd34d376","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 12:58:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40ebf5e86aaf723943876881553488c4fc3d0113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:39:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fc936a6595f48ddb49e3bc83bc0e49b : SUCCESS in 9m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef12ebf9ba924a29b01bebfe9020ae18 : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99c986a83a734b56be8fdf351d91ca15 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8094ba235db947629fba2f0e0f2af274 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6cd0e35d2f843df9b18c4ae578328f6 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78f5a653781d4a8cbbbf13cd0aa364ae : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c1b8baa0b0b43b784c11460cc1bfea3 : SUCCESS in 12m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec01004104af44519a919439d03033db : SUCCESS in 6m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ef0158d8efdc493183139ac9d822dabe : SUCCESS in 6m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/28f6e72f1cb24de4bab0c8eee942521e : SUCCESS in 8m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3a3080dbb9fa4d9c9675a659dda6839c : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dfff36dd0e1845b5a7319e483b930115 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/54495d70c4b946de81c9ad926fe2633d : SUCCESS in 1h 29m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/139bb187dbce434da2f828b054c47405 : SUCCESS in 1h 55m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1fe2ee8453f94585b2c53e46dc987e57 : SUCCESS in 42m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f7faca69ca33426c82c4544592c24a03 : SUCCESS in 36m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/763c1cf1731449f4ac7e8f8877ca2db8 : SUCCESS in 38m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/448e723f24cf494ea301260b7950a356 : FAILURE in 44m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/826cdbf5d06b46cb97822379a323f862 : SUCCESS in 1h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3b4cb8ea75f946eab460b42a148316d8 : FAILURE in 32m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7513551a10b646808a683cea34a38575 : SUCCESS in 1h 50m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e19eafc7685649c5b7a3ac06decf02ae : SUCCESS in 1h 44m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4769136bd60d476db1bc997468fbd801 : FAILURE in 2h 32m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b7ea90bc8d304722b156cc9b24816f55 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"38827e355e1ecb484ca0c43cf6dcf951f353a90a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:41:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bbe21c9e13aadb944792ae3435f19719171bfc1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:13:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0827fc3627954bc3baedd8e58759d54d : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b2871dabc0c45b98cad8c024fa0aa78 : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d09a3c3afae4d07aeeba8e8f326d509 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/515d03d6f92743a59635b109594e0764 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2edaa304e1f42f8b406138d231f53cc : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64141f68e3e74a798b5bb47a8f100ad3 : SUCCESS in 6m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f3f6af61bb40427eadf8b9927037aa59 : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d7b56c2def3443edad8716a696935530 : SUCCESS in 8m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2c45d058473c44fc846332389840ea62 : SUCCESS in 1h 31m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/06795389906d44c9a33637f056bc16f3 : SUCCESS in 1h 21m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1c61771bb7804c5aa8e3700a486c9b8e : SUCCESS in 35m 27s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/69072ce0c1d94c5b8de909620709f184 : SUCCESS in 33m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ddb75699c1484075ab53f6c18785934c : SUCCESS in 33m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d868f25ea1394285a61492f2f2bfbc2f7f77db75","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:13:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d29b51a1aafea3763f4bcb52dc1d923843f18731","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:16:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cee62b0fb234a2888dfd88f6bf3cc16 : SUCCESS in 1m 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2463e0a1e339453d8f3d52849823fbc2 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf73410fdc21fc42df513d79b0f46f8022583a97","revisions":{"bf73410fdc21fc42df513d79b0f46f8022583a97":{"kind":"REWORK","_number":1,"created":"2022-04-12 10:50:14.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/48/837548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/837548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/837548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/837548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/837548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/837548/1"}}},"commit":{"parents":[{"commit":"a16f516d68ad0ed1ccb521ef5b34bf290481c8da","subject":"Merge \"Fix ipv6 interface configuration\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a16f516d68ad0ed1ccb521ef5b34bf290481c8da"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-27 14:19:00.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-12 10:50:10.000000000","tz":120},"subject":"Restart rsyslog from cloud-init","message":"Restart rsyslog from cloud-init\n\nOctavia uploads a rsyslog configuration file using cloud-init but\ndoesn\u0027t ensure that rsyslog is restarted. In some cases (CentOS 9\nStream), rsyslog is started before cloud-init and doesn\u0027t use the\nprovided configuration file.\nThis commit ensures that rsyslog is restarted by cloud-init.\n\nChange-Id: I550c7d2563a2d3ada07e2162383f59f104bcbbd5\n(cherry picked from commit cbffae6eb684629a19844b5f2f4b7523b77a725e)\n(cherry picked from commit 220f13f537cba50f0fb575b0e3b5c39c85e8a0b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf73410fdc21fc42df513d79b0f46f8022583a97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf73410fdc21fc42df513d79b0f46f8022583a97"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
