)]}'
{"id":"openstack%2Foctavia~838438","triplet_id":"openstack%2Foctavia~master~I29fcad9e121a30d6e8f8178f2f078cf10771a32a","project":"openstack/octavia","branch":"master","topic":"v2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-09-14 18:04:39.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-09-02 14:55:47.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I29fcad9e121a30d6e8f8178f2f078cf10771a32a","subject":"Prevent incorrect reschedule of resumed tasks with jobboard","status":"MERGED","created":"2022-04-19 06:48:35.000000000","updated":"2022-09-14 18:06:09.000000000","submitted":"2022-09-14 18:04:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"838438-v2","meta_rev_id":"e5f1bbb90c5fb875818148ee1e9290819d7943b0","_number":838438,"virtual_id_number":838438,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-14 18:04:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-09-14 16:33:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-09-14 07:05:21.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-09-14 16:33:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-19 09:25:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-02 14:48:02.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2022-09-02 14:55:47.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-09-14 16:33:34.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"8edd1caff16b4f52b7cce47f3a3f0ee977950959","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-19 06:48:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebbe6d976478c10f16e7ab26db399d4254c8d8fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 09:25:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55cac95283b942b3bd37aa751d1514c9 : SUCCESS in 12m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd04f51781f141b7b2e96f99b8e79c3a : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de39446036a1496490c5748e8d9f337b : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f2063a6b8fc4e1ebcc035627afd6f30 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3501230b81c487bb6ec3243c1cdfd8a : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c706c8ade4194bdaac9247949613b886 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aff1e6d5e109498588b4472cad6839f5 : SUCCESS in 12m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7d79be1d26fb45539f1828e0cec3803e : SUCCESS in 7m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/236e77aac8404db389c586f560a87281 : SUCCESS in 11m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c07529fed7f14b44bb518fb2cc1d6c2f : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe97e6a78f2a4c6fafc61a5d999e6bd3 : SUCCESS in 9m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2db5a0e0ba614f3f9dc0dff34d788f1f : SUCCESS in 2h 24m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b3dc57916a8a4cadaa5596cb54de6c16 : SUCCESS in 1h 44m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/979c8fba03ad4d4eb4f6f612e8beb3d2 : SUCCESS in 37m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e58061d154c5489ab9b52c6f2a961c42 : SUCCESS in 1h 07m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bdf94430da6d449fb3803bbf70863e13 : SUCCESS in 57m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b08c61e0badf4ecfa1d00db203c8dd76 : SUCCESS in 38m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ed16382efe0941d49a522a151a287113 : SUCCESS in 1h 29m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f6b554c757f5436b979dac617922f482 : SUCCESS in 2h 07m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/65667151e89448aaa6bb36a01adeea04 : SUCCESS in 1h 33m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0a336b84779e4c4280ca9ff8283f139e : SUCCESS in 2h 03m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bd22162ebc49458d837c980bb6cd7a66 : SUCCESS in 1h 32m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab3f79338bd34e308ce59cd6a2100d95 : TIMED_OUT in 2h 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df4d8e93440837afc112cfba986d10494dcf37d5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-19 13:31:59.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b1249655b13d93b32db3165c82e382fa82229181","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-02 14:15:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3f87e328818fc3420075bedde243b90d0f869d5b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-02 14:15:55.000000000","message":"Patch Set 2: Backport-Candidate+2","accounts_in_message":[],"_revision_number":2},{"id":"d485b9870833d6f998266ceb1273bc2383188742","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-02 14:48:02.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0385a0d3988eba3fdcf063dc286de910ede232e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-02 14:53:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c342b9c9af4e5106397a0ff6882215ea3c357aa9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-02 14:53:41.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6e953fda3015855cfb8ee39d24466299ec41f6de","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-02 14:55:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2e46960b5af89926840a6bd972cf2301852b1699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 17:50:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fbb3261f7a9441386b58d99b4555757\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bc088227d264e6085e83fdd01525c82 : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acf9e4b7be6f4744ac4585f71bac4088 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/943d9f2a8bc34e59bf417c944cdd92c3 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77494a907fcf482c8f7d4e257076a07b : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ea5b86be5364ec593ae9b3ad6adda8f : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4126714d81f849ab88e1fe86e8ec6c29 : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e119dc9762474398b65bb4da8cb70627 : SUCCESS in 5m 28s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5ae8fc129d3d4bc18c470ca50aa00bab : SUCCESS in 5m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d9f3c46dfe8c4c42809314c99ac7c97e : SUCCESS in 8m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9e409df74fae4eeea926e1d1ed699c7a : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/49f415233cc04c77a725fcbf317ee562 : SUCCESS in 7m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/94382b9bfe6246a5b109577ed8ab5c0a : SUCCESS in 7m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/70ceefa5fcb748cab55db8eabd3a997a : SUCCESS in 7m 55s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/308c4a8e2f9e47eebef84cc0e94206cc : SUCCESS in 2h 49m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ebdfa49e5ea94c188a1b95834b0b0845 : SUCCESS in 2h 06m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6dcab700474147ecacc5b01a1bcaaec9 : SUCCESS in 44m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b2d30517c9db43ea8968c5037c74006f : SUCCESS in 1h 12m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/860f8e57b3504634b11765ad32527bd7 : SUCCESS in 1h 18m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c6ad4ebbd674db6a85175e210b6c8a2 : SUCCESS in 40m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cc99587be3be4058a4257feba1708493 : SUCCESS in 1h 57m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8ae0730568954de29f967979b808068a : SUCCESS in 2h 07m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4bd6a6526bf845c6b2926af1862f13ae : SUCCESS in 2h 02m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/58b67cf1a8e14a7ab6490e59a2f745fa : SUCCESS in 2h 07m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/56c3d7e9d4cf4c92a0163f1441787b94 : SUCCESS in 2h 00m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc23089c2ad34410a69f5c3304a67bf9 : SUCCESS in 2h 05m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e33f7c6b8065ed3d193de7baadde9aee8f24c61d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-06 12:58:06.000000000","message":"Patch Set 3: Backport-Candidate+2","accounts_in_message":[],"_revision_number":3},{"id":"47e860e3dececc273db94bb8182f0a4246418421","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-13 06:24:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"415dc64315b5bd7baf6b848bd492ec3249b14983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 09:44:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ef95124e0df47349fb35ffd2f6c3190\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfd2f59f2df34e668292b3ac1ff7d3ff : SUCCESS in 12m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c164e96602314320b54205526f63b0b5 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47b94b102a04468eb9bb209adee77e7f : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/debffad55cf248a68d80bde8f02c6fd3 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69fffbd60b4f4071b5c065d5505d8088 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d579d8626974afcbabc2699b09cf220 : SUCCESS in 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea1018221f354711a87100f1ed5aaf48 : SUCCESS in 4m 28s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/dc25e21f5d484e96a1be3dfbe2d5e141 : SUCCESS in 4m 53s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/cb70a3a92089430fa09f5026e57420d3 : SUCCESS in 8m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3f56ac55021f4e9382363467add3e145 : SUCCESS in 4m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9d0d749f5df9487195ac919eb9960cb8 : SUCCESS in 6m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/124f24e11cf741b0a02be7bbbe00ddc7 : SUCCESS in 8m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/36be199375c64cb7953cb64e4eb54a5e : SUCCESS in 7m 10s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/92434dae98414b98ba257c25cb347577 : SUCCESS in 1h 35m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54c62f3f89f94c5983d53c0acb43ee01 : SUCCESS in 1h 59m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b358b9614fee4616a60629439f99b621 : SUCCESS in 47m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a180aa1fbf224359bc27e6d85bdcbb76 : SUCCESS in 57m 28s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ada2d39bb8554893b8c891f90eceeac3 : SUCCESS in 1h 30m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/76d3e02cc67449c78ebc861ad8997838 : SUCCESS in 38m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/499ae2530d4c411f88d81c4a1a5894e2 : SUCCESS in 2h 26m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7b7de229d98f4155bbe490637f34d8c3 : TIMED_OUT in 3h 09m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/527d08fbfc8041a9b32134fd9b424049 : SUCCESS in 1h 50m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ac3351be49604a648108668d6d74671d : SUCCESS in 1h 50m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/82b6ece64ce44b6495164907479de31f : SUCCESS in 1h 47m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/78aaaf1813b34cb7949ebc0e6a75cf43 : SUCCESS in 2h 25m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ed0cc6d8120715ff2ea329d9299c71e9316e35af","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-14 07:05:21.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9895db2ff786aea21065a042da1bb2739304c0b0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-14 16:33:34.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8b2d2918bc72cfa667ef064f1acfd5012223dd28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 16:34:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8bcb51959ee240d4b2e24aa399373d0cab5c4d36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 18:04:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ed86a17d9e647c08e79cd3abb2470f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b62b7570a24402db6ca88c8f4c1c079 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64d96c6533d74d49a2699026950b39c1 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ae108339eef4b2090920c89bf007b4c : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d092c3f07b414cc0bb4b140940127586 : SUCCESS in 12m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6311b29874f14a1993e6bd3b10c4b45b : SUCCESS in 6m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/38d7499cb00f4b3ca536c8e021d9ce47 : SUCCESS in 5m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b9ef4d7456ea45b9aead1ae48ea4f60e : SUCCESS in 7m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a3623856cbb849e3834a8f76b28795e3 : SUCCESS in 1h 28m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f4ace93c2ac450d920101928630fd49 : SUCCESS in 1h 29m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/603e066715e74cf4a4748cd07f5f96af : SUCCESS in 40m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7753443104024acaaf90bf1f9c220a00 : SUCCESS in 53m 50s","accounts_in_message":[],"_revision_number":4},{"id":"689ae98e22577ff3e75d2e0baf2354e4da99bb0a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 18:04:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e5f1bbb90c5fb875818148ee1e9290819d7943b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 18:06:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bc030e1bb8b461eb8e9d648f6c5fe64\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bd37eda242949f29c2d49c1c7f6fa6f : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fe4cb509f4b4cfbba3bab201eb4f256 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d367b47639af02aaaff80cd62629f5c8e7f3f91a","revisions":{"eb63883898e71773d6fea81987c8f2c303af5d14":{"kind":"REWORK","_number":1,"created":"2022-04-19 06:48:35.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/38/838438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/838438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/838438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/838438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/838438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/838438/1"}}},"commit":{"parents":[{"commit":"bf007ec4a8bf368321b66038388a6bf67c5d22ce","subject":"Merge \"Add a Grafana dashboard for Octavia load balancers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf007ec4a8bf368321b66038388a6bf67c5d22ce"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-15 12:53:05.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-19 06:47:50.000000000","tz":120},"subject":"Prevent incorrect reschedule of resumed tasks with jobboard","message":"Prevent incorrect reschedule of resumed tasks with jobboard\n\nA previous patch [0] introduced a keepalive mechanism that allows a\ncontroller to notify taskflow that a task is being executed.\nBut it doesn\u0027t handle \"resumed\" tasks (tasks that have been rescheduled\nfrom one controller to another controller).\n\nThis patch moves this keepalive task in a thread that is attached to a\ntaskflow conductor, the thread will extend the expiration of the tasks\nthat are executed by this conductor.\n\n[0] I6619ba117f7051fe44086389789bc6e2810fd23d\n\nStory 2009998\nTask 45089\n\nChange-Id: I29fcad9e121a30d6e8f8178f2f078cf10771a32a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb63883898e71773d6fea81987c8f2c303af5d14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb63883898e71773d6fea81987c8f2c303af5d14"}]},"branch":"refs/heads/master"},"bb34dab3633eaba7b4b3bb5fb1a4c9b96980826f":{"kind":"REWORK","_number":2,"created":"2022-09-02 14:15:06.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/38/838438/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/838438/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/838438/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/838438/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/838438/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/838438/2"}}},"commit":{"parents":[{"commit":"8c793f2e8ff7857a97cc68a4706bdcdfdff1fde5","subject":"Merge \"Allow multiple VIPs per LB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c793f2e8ff7857a97cc68a4706bdcdfdff1fde5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-15 12:53:05.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-09-02 14:14:40.000000000","tz":120},"subject":"Prevent incorrect reschedule of resumed tasks with jobboard","message":"Prevent incorrect reschedule of resumed tasks with jobboard\n\nA previous patch [0] introduced a keepalive mechanism that allows a\ncontroller to notify taskflow that a task is being executed.\nBut it doesn\u0027t handle \"resumed\" tasks (tasks that have been rescheduled\nfrom one controller to another controller).\n\nThis patch moves this keepalive task in a thread that is attached to a\ntaskflow conductor, the thread will extend the expiration of the tasks\nthat are executed by this conductor.\n\n[0] I6619ba117f7051fe44086389789bc6e2810fd23d\n\nStory 2009998\nTask 45089\n\nChange-Id: I29fcad9e121a30d6e8f8178f2f078cf10771a32a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb34dab3633eaba7b4b3bb5fb1a4c9b96980826f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb34dab3633eaba7b4b3bb5fb1a4c9b96980826f"}]},"branch":"refs/heads/master"},"cde504cf8446d7b3027405f28ecbe12bc993c9a4":{"kind":"REWORK","_number":3,"created":"2022-09-02 14:53:16.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/38/838438/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/838438/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/838438/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/838438/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/838438/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/838438/3"}}},"commit":{"parents":[{"commit":"8c793f2e8ff7857a97cc68a4706bdcdfdff1fde5","subject":"Merge \"Allow multiple VIPs per LB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c793f2e8ff7857a97cc68a4706bdcdfdff1fde5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-15 12:53:05.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-09-02 14:52:14.000000000","tz":120},"subject":"Prevent incorrect reschedule of resumed tasks with jobboard","message":"Prevent incorrect reschedule of resumed tasks with jobboard\n\nA previous patch [0] introduced a keepalive mechanism that allows a\ncontroller to notify taskflow that a task is being executed.\nBut it doesn\u0027t handle \"resumed\" tasks (tasks that have been rescheduled\nfrom one controller to another controller).\n\nThis patch moves this keepalive task in a thread that is attached to a\ntaskflow conductor, the thread will extend the expiration of the tasks\nthat are executed by this conductor.\n\n[0] I6619ba117f7051fe44086389789bc6e2810fd23d\n\nStory 2009998\nTask 45089\n\nChange-Id: I29fcad9e121a30d6e8f8178f2f078cf10771a32a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cde504cf8446d7b3027405f28ecbe12bc993c9a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cde504cf8446d7b3027405f28ecbe12bc993c9a4"}]},"branch":"refs/heads/master"},"d367b47639af02aaaff80cd62629f5c8e7f3f91a":{"kind":"REWORK","_number":4,"created":"2022-09-13 06:24:20.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/38/838438/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/838438/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/838438/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/838438/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/838438/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/838438/4"}}},"commit":{"parents":[{"commit":"8caca9219f36c32a4ed7fd28d03206a49a15a083","subject":"Merge \"Add missing parameter in amphorav1 batch member update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8caca9219f36c32a4ed7fd28d03206a49a15a083"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-15 12:53:05.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-09-13 06:24:09.000000000","tz":120},"subject":"Prevent incorrect reschedule of resumed tasks with jobboard","message":"Prevent incorrect reschedule of resumed tasks with jobboard\n\nA previous patch [0] introduced a keepalive mechanism that allows a\ncontroller to notify taskflow that a task is being executed.\nBut it doesn\u0027t handle \"resumed\" tasks (tasks that have been rescheduled\nfrom one controller to another controller).\n\nThis patch moves this keepalive task in a thread that is attached to a\ntaskflow conductor, the thread will extend the expiration of the tasks\nthat are executed by this conductor.\n\n[0] I6619ba117f7051fe44086389789bc6e2810fd23d\n\nStory 2009998\nTask 45089\n\nChange-Id: I29fcad9e121a30d6e8f8178f2f078cf10771a32a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d367b47639af02aaaff80cd62629f5c8e7f3f91a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d367b47639af02aaaff80cd62629f5c8e7f3f91a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
