)]}'
{"id":"openstack%2Foctavia~838503","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I888a672b876e2914c56f882f8bb8cb67721506cf","project":"openstack/octavia","branch":"stable/yoga","hashtags":[],"change_id":"I888a672b876e2914c56f882f8bb8cb67721506cf","subject":"Fix grenade job by using dib git repo","status":"ABANDONED","created":"2022-04-19 12:33:21.000000000","updated":"2022-04-27 12:56:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"36d13e48cde9e333c78730e60b3e2203ceabd40c","_number":838503,"virtual_id_number":838503,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-19 15:12:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-04-19 15:12:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"74ec19d1809393382420b3f42e3fcc41b4073eeb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-19 12:33:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e60faf2533150f39243a318fc693532f198e4741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 15:12:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f82d93c07854263abb40c0f4a48177f : SUCCESS in 12m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/406050d80e8146cf854ee3bd499d5d88 : SUCCESS in 14m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6011f79124f34c858cfaa958b90302dc : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4072f66bade246fab1e457e298ae8789 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adaa57968ffa4ceca5174fd1fdbd8a64 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08202c8dbde04d888b58596affbce630 : SUCCESS in 11m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f4473855a4b54946a8778175cc225c3e : SUCCESS in 6m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7977076c80cc43faaa6973f17b3d75ae : SUCCESS in 11m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/883725216cc340c99fef7df9f0a84d2b : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e40a6df5f6e4bf09bbc30d066914ae1 : SUCCESS in 8m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df9359c90af442fda14c2ced82c819dd : SUCCESS in 2h 01m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5498d83523c14406961c42116e51c6a2 : SUCCESS in 1h 20m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ab060502f21f4d7aa51e841abab25fac : SUCCESS in 35m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b27fb39d409480c96cc519a5c921371 : SUCCESS in 48m 47s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6474f66770bc40e0a293118faa42748d : SUCCESS in 57m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e95f2fa106f470cb287048271a45a71 : SUCCESS in 33m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/11794af3dcc541b3bb44b11d91e67d50 : SUCCESS in 1h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8c938b13e57944238781dcc9f19f6146 : SUCCESS in 1h 50m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3f5c247e0a5b4c46a3b62385c9803f4e : SUCCESS in 1h 45m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/154a18c01be64f1499e017a1fb46003c : SUCCESS in 1h 45m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9feda7a18a01421e97830628e9c74ef7 : POST_FAILURE in 1h 53m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/db79151d37934868beb344e1e305c6ed : TIMED_OUT in 2h 34m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"36d13e48cde9e333c78730e60b3e2203ceabd40c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 12:56:28.000000000","message":"Abandoned\n\nwas fix with an update in requirements","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0026b9978a776f7ded28e8d14b41db763ecc7877","revisions":{"0026b9978a776f7ded28e8d14b41db763ecc7877":{"kind":"REWORK","_number":1,"created":"2022-04-19 12:33:21.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/03/838503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/838503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/838503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/838503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/838503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/838503/1"}}},"commit":{"parents":[{"commit":"a8a35e088cc28fcbd3ed7b054b7ef97ba255228a","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8a35e088cc28fcbd3ed7b054b7ef97ba255228a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-15 06:20:35.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-19 12:33:02.000000000","tz":120},"subject":"Fix grenade job by using dib git repo","message":"Fix grenade job by using dib git repo\n\nA recent git update broke our jobs, diskimage-builder was updated to fix\nthis issue but the grenade job doesn\u0027t use dib from master (our other\njobs do).\nThis commit ensure that our grenade job pulls dib from the git repo.\n\nChange-Id: I888a672b876e2914c56f882f8bb8cb67721506cf\n(cherry picked from commit b14a349e1426df9ed4536aa55860446f99537ae6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0026b9978a776f7ded28e8d14b41db763ecc7877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0026b9978a776f7ded28e8d14b41db763ecc7877"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
