)]}'
{"id":"openstack%2Foctavia~838504","triplet_id":"openstack%2Foctavia~stable%2Fxena~I888a672b876e2914c56f882f8bb8cb67721506cf","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-04-27 12:56:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I888a672b876e2914c56f882f8bb8cb67721506cf","subject":"Fix grenade job by using dib git repo","status":"ABANDONED","created":"2022-04-19 12:33:51.000000000","updated":"2022-04-27 12:56:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"87d2bbafb601c615772ee63b9be13b054feae411","_number":838504,"virtual_id_number":838504,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-19 15:23:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-20 08:35:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"reviewer_updates":[{"updated":"2022-04-19 15:23:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-20 08:35:24.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"9b76fc4b07d8528516719fb52296512acdcf11f2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-19 12:33:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6d7ac2d6cf7e337683c1fe881d1b98d3a6781de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 15:23:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34b0ce38939c4d3baf1b68ac4eaf03ca : SUCCESS in 12m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfba52f242364e80a8581f142e81a914 : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69cd02daba704100b90a09805a402342 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ceaa240c4144882b835c094fcf7ef67 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01b9fccf8d9648b5bdef2b63e91b5580 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b8b9795c2664a5bbde921d0e0e05943 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/739c3d88af1849b28f7af848c6da62e4 : SUCCESS in 11m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/70092463b169415989ab93f102e17476 : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/96753968d364478fab54fcdde6366533 : SUCCESS in 10m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cd4c4dae723a425caa8a2e2953fd0106 : SUCCESS in 8m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/66443b22965645949928fb0a7b235874 : SUCCESS in 8m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/90c4fc424e56473799b1534eb138c0a1 : SUCCESS in 2h 02m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7123a9d35ae442f893debccf58ddd46e : SUCCESS in 1h 31m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7e611ab6482c4f9b807ae80b3434d379 : SUCCESS in 37m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9a08cec2736f4fc19d8f1ee84ff20152 : SUCCESS in 1h 04m 33s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/51293d7cecda4524a949f7818d8db334 : SUCCESS in 2h 09m 38s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1f1d8dcbfce048708e3d01b6d4e2c56f : SUCCESS in 46m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9243e31fa269491a9e6cabedce5afecf : SUCCESS in 34m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8a6a2bb2bd464daf82f027f0cf6e4752 : SUCCESS in 1h 22m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5223a34dd2a34b46abe88ced7e89d35f : SUCCESS in 1h 47m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4d819198c6d041e7a1c1f57135fb2b92 : SUCCESS in 1h 35m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4ec13bd19bdb4595ba1935a0e2e95055 : SUCCESS in 1h 37m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6a6977d6a9ef4d4d8c59be88fe722cbc : SUCCESS in 1h 19m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d26ffe3d4cb24042a76ec89f0c10b868 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"55d09888f09d5586ccad5057f5b89a10fa92dd46","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-20 08:35:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"87d2bbafb601c615772ee63b9be13b054feae411","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 12:56:23.000000000","message":"Abandoned\n\nwas fix with an update in requirements","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b22934b8bc6b76d282093e4c9e9291c93062f458","revisions":{"b22934b8bc6b76d282093e4c9e9291c93062f458":{"kind":"REWORK","_number":1,"created":"2022-04-19 12:33:51.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/04/838504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/04/838504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/838504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/838504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/838504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/04/838504/1"}}},"commit":{"parents":[{"commit":"27a8cc7e15b4a5f6e602b903b77a92a98d9a413c","subject":"Merge \"Add keepalive for redis-based taskflow boards\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27a8cc7e15b4a5f6e602b903b77a92a98d9a413c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-15 06:20:35.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-19 12:33:45.000000000","tz":120},"subject":"Fix grenade job by using dib git repo","message":"Fix grenade job by using dib git repo\n\nA recent git update broke our jobs, diskimage-builder was updated to fix\nthis issue but the grenade job doesn\u0027t use dib from master (our other\njobs do).\nThis commit ensure that our grenade job pulls dib from the git repo.\n\nChange-Id: I888a672b876e2914c56f882f8bb8cb67721506cf\n(cherry picked from commit b14a349e1426df9ed4536aa55860446f99537ae6)\n(cherry picked from commit 0026b9978a776f7ded28e8d14b41db763ecc7877)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b22934b8bc6b76d282093e4c9e9291c93062f458"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b22934b8bc6b76d282093e4c9e9291c93062f458"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
