)]}'
{"id":"openstack%2Foctavia~838732","triplet_id":"openstack%2Foctavia~stable%2Fxena~I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-04-25 14:17:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081","subject":"Save the HAProxy state outside of its systemd unit","status":"MERGED","created":"2022-04-20 14:35:19.000000000","updated":"2022-04-25 14:19:43.000000000","submitted":"2022-04-25 14:17:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838732","meta_rev_id":"5ab368818cc3bc94f9b861bef8b139ec20445d73","_number":838732,"virtual_id_number":838732,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-25 14:17:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-25 08:36:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-04-25 11:51:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-04-25 11:51:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-20 14:35:19.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"CC"},{"updated":"2022-04-20 16:48:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-25 08:36:14.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-04-25 11:51:13.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"2122a8f4a1faf52d860892fbaf05b0d5d7eef8d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-20 14:35:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad2a2f5f333d2aae6f81a8aa74be47bf014a9093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 16:48:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52e510b294954765b052d4fd2b29461d : SUCCESS in 10m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fc87f75bb344a298ab53e0ea4a20c8a : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e797608610944eb8faac5af0d659f10 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84229cdc8ddc44e3b291ccbcb17d8137 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/705025096f5d4fbfaf53691d679570fb : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fcd7a33ed264011a7cbd0cbd6cf89c4 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79a1cb83604a4e79b507e517eb97d77f : SUCCESS in 12m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b6b6819cfd84572a2720dbd07726df4 : SUCCESS in 5m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/58bdfb3fc828490598009b1a1cd6d7df : SUCCESS in 6m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ec9a6a6605e744fc8a693e1fab51e394 : SUCCESS in 8m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7eb9f3eaa7784ea2b238ce3b56d660fc : SUCCESS in 5m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dd2cdb811c8b4c148a6bbca2a467e4bc : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/617d752801244dbdaa7bb783c997c9c7 : SUCCESS in 2h 06m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7df9037d68b40ce82efa719884f15f0 : SUCCESS in 1h 39m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2319b69665cc4fa2ad7f91b79b04e047 : SUCCESS in 38m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/52234782caa942b6b312781a11bc0e73 : FAILURE in 23m 02s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/2420ccd9f8cf4080a27e873ca45a6ffb : FAILURE in 13m 11s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/39b9d1c6a41d4fccaf6b626c8a181565 : FAILURE in 14m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9866621dded0482bb36af332a2733729 : SUCCESS in 33m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e8ecff33aa8f49cf893f178fe67c1526 : SUCCESS in 1h 55m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/38af526d53904e45b6ab876ad327bb4e : SUCCESS in 2h 05m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/51ae2699cb154b9097f232ef6ac86198 : SUCCESS in 1h 51m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d22ce9443f344aa39ba5f5b560588257 : SUCCESS in 1h 38m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b5c195631f4e4000ac7e849e31372458 : SUCCESS in 2h 02m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7075d44e630344499a126a1633836560 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"815ea4e46cd46e794979e1b895273c34d603b08f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-21 08:01:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef36a87bf97d10c75460eb0d90f05372108a0c5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 10:31:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02c10375d3424834acb369c15c77c89b : SUCCESS in 11m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cd442616aa54bd9903a2f3aeb0ede52 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60a6d9ed878943fbab77d5105c5dd7ce : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f4572d1956c4a488cc12a8e3fe7715d : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d26f3a222c89403682114f0bd34a3406 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48dfea03ac2545dfb08249d5597125fe : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88c63535eb9e4fa6889484cdf6067af0 : SUCCESS in 12m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4419f1c1eae4731abe73401357de340 : SUCCESS in 5m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bdf3279b15ea4e17b9c2e5921a923151 : SUCCESS in 6m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b8c6113fda4742688f14d5f5761a0471 : SUCCESS in 9m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6ef6cab0cc054ec5ab9a7762f803c430 : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2dc5c3ee787a479883133bc13623dfa1 : SUCCESS in 9m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/377562e512e54bc79df66a74feca0a05 : SUCCESS in 2h 27m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/878f1ec9d6bf412e93c294b5132c638a : SUCCESS in 1h 24m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f12308cc096146b39e7117689b60e3a7 : SUCCESS in 37m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d9156e76df3b45578ef7db284b0611c1 : SUCCESS in 49m 24s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/bbeb7bb356694b2aa884f5767e80e104 : SUCCESS in 2h 07m 03s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1f82a8394c72462eab6d993e411c568b : SUCCESS in 55m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d1c8d4beefe4fdebdf63c021d3fb9a1 : SUCCESS in 35m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f078e68b57bc43ce83f00ca8a9f56ba6 : SUCCESS in 1h 31m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c0210c7188fe4c6985f5c3008eb41bfb : SUCCESS in 1h 42m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ff7fe233c3d04a0dbd822a78a9986261 : SUCCESS in 1h 34m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/67ae9565d05f480fad706c963076c85d : SUCCESS in 1h 42m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e3a7e26cf0cb4648a774c1bd5f971bea : SUCCESS in 1h 30m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4e81eced5df54b979fa6adeeb2118d9b : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"20764504a66d3f8684e93337732d25060d1163de","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-25 08:36:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1d063884806c7daddd44dac9407564faf73591ba","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-25 11:51:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e21325c3ff7b13f427c1a0d3b687704e2bf9b5c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 11:52:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"339c2ee95e9d65fb42315861e164a39cb32a6c44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:17:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2300a6d7041d4fbaa8b12a7bb95b1c43 : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/810952ae3c9f48b5839d23f3de129407 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0d31920a256424b809506e51abd3012 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/449554a8aaf2433399c00438ca670c8b : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fcabc36e1a5485bb2ffc14679234c04 : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d32b5dc84db94232aafd294d13266dc4 : SUCCESS in 5m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/eb1899f6443f45608cdf43eb806ff9e2 : SUCCESS in 6m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a6013fa99dd40298b9da551ea500e96 : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d10fae03e7304dc989a2345bf5b11ee8 : SUCCESS in 1h 43m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd3d5e809602406c8f7275bc9966aa1d : SUCCESS in 1h 41m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/50f59c78eee64453990cb650020b463f : SUCCESS in 31m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/62794f34d8b145a7bb8437572750b546 : SUCCESS in 50m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6580cff1c7bad3c7717e0525dfb318916126126b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:17:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5ab368818cc3bc94f9b861bef8b139ec20445d73","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:19:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a24293db08d646be8fc70556d718330a : SUCCESS in 1m 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7efc464fd6c49768aa94548385c3ac5 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"edcd6931fcc83dffc4d4a37c91b8db044ff3fdc4","revisions":{"edcd6931fcc83dffc4d4a37c91b8db044ff3fdc4":{"kind":"REWORK","_number":1,"created":"2022-04-20 14:35:19.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/32/838732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/838732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/838732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/838732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/838732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/838732/1"}}},"commit":{"parents":[{"commit":"27a8cc7e15b4a5f6e602b903b77a92a98d9a413c","subject":"Merge \"Add keepalive for redis-based taskflow boards\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27a8cc7e15b4a5f6e602b903b77a92a98d9a413c"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-04-13 11:52:53.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-20 14:35:01.000000000","tz":120},"subject":"Save the HAProxy state outside of its systemd unit","message":"Save the HAProxy state outside of its systemd unit\n\nBy default, SELinux prevents HAProxy context (haproxy_t) to execute\nshell context (shell_exec_t) for security reasons.\n\nThis prevents HAProxy to actually reload properly, since SELinux will\ndeny its call to a shell to save its state to a file.\n\nIn order to avoid opening a potential security hole in the load-balancer\nimage, the best way is to generate the state file before the actual\nreload.\n\nThere are more details about the SELinux denials in the associated Red\nHat Bugzilla.\n\nResolves: rhbz#2073491\nChange-Id: I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081\n(cherry picked from commit 21d74c373b92f8711ff91766c2e7c8bcb811bdda)\n(cherry picked from commit 0a062cd66416c6020d9718d87988727bb8a78627)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/edcd6931fcc83dffc4d4a37c91b8db044ff3fdc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/edcd6931fcc83dffc4d4a37c91b8db044ff3fdc4"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
