)]}'
{"id":"openstack%2Foctavia~838733","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-04-25 14:17:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081","subject":"Save the HAProxy state outside of its systemd unit","status":"MERGED","created":"2022-04-20 14:35:53.000000000","updated":"2022-04-25 14:19:51.000000000","submitted":"2022-04-25 14:17:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838733","meta_rev_id":"95dd55e82148968e6634c55b8d3bd9559622452a","_number":838733,"virtual_id_number":838733,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-25 14:17:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-25 08:35:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-04-25 11:54:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-04-25 11:54:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"CC":[{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-20 14:35:53.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"CC"},{"updated":"2022-04-20 16:26:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-22 06:32:15.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2022-04-25 08:35:43.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-04-25 11:54:42.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"12c43f5688034c4d33984a7159fe49f2c5406c5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-20 14:35:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"722880173b89ea06db273c3a02896659891bf1d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 16:26:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3347ed39a1334b088cef7b1063edcfda : SUCCESS in 10m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d088c59d84f43dda97aabdf1124b65e : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12ed70cb3adc454c9ee10abcb8c30085 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f264edb31a841d68ab67a70b1e959db : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b081a38415d94a9aa59e45affbb4580f : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b158c49358f419fa1c2ca441d9cbf51 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a191f41842e42a5ba0efcb9cd32d526 : SUCCESS in 13m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57f80b8996174b6d8937f1719e0524b5 : SUCCESS in 5m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/85b6f49eed764444a1595b34c9d7a451 : SUCCESS in 6m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/99db536f29e44ebd84eccd8f7b1acbcd : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/756ab02c1730459c93f2b92977e4309a : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/23bae8c468b4446b9dac37eabd3ef5ba : SUCCESS in 9m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/998e7eef8e5d48bd97408c9dbfd80fac : SUCCESS in 1h 48m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9444e83ddcf84f66b546ab3da0e2540f : SUCCESS in 1h 18m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3a8b796a29bd42469d4c2d49403505e3 : SUCCESS in 36m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7c6a6aac7fe84ae8a0e242e1759d5177 : SUCCESS in 36m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7134a6b4657948ebbb4b9b6120561954 : SUCCESS in 35m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5a758ae2ec1140b49193ccfae645073f : FAILURE in 25m 26s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fa144c45844f42c0bc7f49efb9402490 : FAILURE in 13m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d9e956300227474bb96ed28587a8722d : SUCCESS in 1h 25m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/16d796c4fe924d0eb51573eebba0af5f : FAILURE in 33m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/66022836d5c949229e0459a33428c3ae : SUCCESS in 1h 44m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7486dce4b5844176943afc5bde0efe61 : SUCCESS in 1h 46m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2a54ab2c1c894b789150e51536ac3a6e : SUCCESS in 1h 31m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7471e6d7097e436aa2196b4880f8bd5a : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a6d861897a62808d18c455660c6a5fa99329b58b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-22 06:32:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4dc2a42a5d8875b85e2520351b4953e81b4a6438","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 08:24:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed8cc8afb4c64089b9d251bff697fb51 : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de671a9546d24da1949cfa493b13289f : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57cbd0e853e402a81bcf0a58d58ea34 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ec57fae2f2c4331b7e44ae3ebbde415 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d99c58489e1b4eae8c9c4866dd408bc6 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/913ad5edaf374386a63b5b2b901758bd : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53f30de96ead4961918c683f8224f732 : SUCCESS in 13m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62f79d25d3ac47e4bae8b2b7ceaf2f81 : SUCCESS in 5m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f523be9a8ce74c9ab67a7c5363619047 : SUCCESS in 5m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/793604149815481dbee197cd36dc3898 : SUCCESS in 8m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/706dcd85c3924c50b845c41f3db2c86f : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d1e48d1dd0704f229e17dc69830ad512 : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/88b09391f69d4e4fa3dd64d0d8ecc26f : SUCCESS in 1h 31m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d67b7bcae8864b6aa9a796864ad29ad5 : SUCCESS in 1h 21m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d2108ae70a8d4abaaadda973ad209832 : SUCCESS in 41m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/569e7af250b5407c8002e961cb18d003 : SUCCESS in 30m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/24f657515bc44ea4be7be4a17cf2341a : SUCCESS in 32m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/154e70833c3c4b7a96186dbc1d6a4702 : FAILURE in 41m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4cbdb3e7269a44568dccff4a87e11ccb : SUCCESS in 1h 43m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/deb43dbc236d4e4fbc71ac580d271167 : FAILURE in 32m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7e8c2824af4d4a9b94d511dc36777e68 : SUCCESS in 1h 26m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b62bfc17aef9450eafd9a5a6a65a03ab : SUCCESS in 1h 28m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/98f9be8c4a8347d4831fd2187d629b15 : SUCCESS in 1h 23m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6fa25694ef644b72ae20d8a80c2a82ff : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26f613d9b8cec83ed590d412ed894cd7dc31220b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-25 08:35:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"332b209a4e9a9d4aa5ca502838b7116e7619562f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-25 11:54:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f9fc8deb219497f034d8b461e5ba3c01422969a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 11:55:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93fc7f7b2552dc3b0a47200e0647d4b2edcacd6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:17:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9dfc7d28a6754059bae5196de393ebac : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/921734810820496483660a7a8ee8d732 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d3c3f7974b640358693673191221ddc : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/221ee6da36494b70adb8f9edb632ae97 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b655d2599b74f7cb190d8693a1d41e7 : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d448a3dd5efb4a28b61feb3fe5170c34 : SUCCESS in 4m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/db4ec20e689442aa8166fd25b68076dd : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f9ba129a976642edb5881a7203af3623 : SUCCESS in 6m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/437a0751aef3485b8b3ab1f33ae5d9f8 : SUCCESS in 1h 37m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a2e4fb12bdf4fcdab89d98bc7011fe8 : SUCCESS in 1h 22m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d05b5f0d41074b3294514c170c4a4138 : SUCCESS in 36m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3000c77f682f481f8098d8a9332dde51 : SUCCESS in 29m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a3930735c8954d87abfa445b46dab85d : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":1},{"id":"814eb11268c8805b74e261fc33cdaa2dbc8fedae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:17:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"95dd55e82148968e6634c55b8d3bd9559622452a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:19:51.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/264125d2a6d043c5a151610e22c77294 : SUCCESS in 1m 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e0e27e484314a56828ab59766435aa7 : FAILURE in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97ac37bc6e6572ce7b368c974b928ce5e35700d1","revisions":{"97ac37bc6e6572ce7b368c974b928ce5e35700d1":{"kind":"REWORK","_number":1,"created":"2022-04-20 14:35:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/33/838733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/838733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/838733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/838733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/838733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/838733/1"}}},"commit":{"parents":[{"commit":"1329b5799e846fdedd115d75a58d1be4157a092c","subject":"Merge \"Use centos amphora image in the FIPS jobs\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1329b5799e846fdedd115d75a58d1be4157a092c"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-04-13 11:52:53.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-20 14:35:20.000000000","tz":120},"subject":"Save the HAProxy state outside of its systemd unit","message":"Save the HAProxy state outside of its systemd unit\n\nBy default, SELinux prevents HAProxy context (haproxy_t) to execute\nshell context (shell_exec_t) for security reasons.\n\nThis prevents HAProxy to actually reload properly, since SELinux will\ndeny its call to a shell to save its state to a file.\n\nIn order to avoid opening a potential security hole in the load-balancer\nimage, the best way is to generate the state file before the actual\nreload.\n\nThere are more details about the SELinux denials in the associated Red\nHat Bugzilla.\n\nResolves: rhbz#2073491\nChange-Id: I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081\n(cherry picked from commit 21d74c373b92f8711ff91766c2e7c8bcb811bdda)\n(cherry picked from commit 0a062cd66416c6020d9718d87988727bb8a78627)\n(cherry picked from commit edcd6931fcc83dffc4d4a37c91b8db044ff3fdc4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97ac37bc6e6572ce7b368c974b928ce5e35700d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97ac37bc6e6572ce7b368c974b928ce5e35700d1"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
