)]}'
{"id":"openstack%2Foctavia~838736","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081","project":"openstack/octavia","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-04-25 14:17:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081","subject":"Save the HAProxy state outside of its systemd unit","status":"MERGED","created":"2022-04-20 14:41:56.000000000","updated":"2022-04-25 14:20:01.000000000","submitted":"2022-04-25 14:17:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838736","meta_rev_id":"ad05c23c7f68c753ed4e571ecde71347b00da2af","_number":838736,"virtual_id_number":838736,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-25 14:17:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-25 08:36:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-04-25 11:55:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-04-25 11:55:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-20 14:41:56.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"CC"},{"updated":"2022-04-20 16:43:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-25 08:36:33.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-04-25 11:55:48.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"a6886693cbc4f1d24b3e6699cab6949f71445984","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-20 14:41:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69c498841c461a28b63fa394bb14cb3fe1248f2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 16:43:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/671513a780104cebbc90f59286a5b4e8 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bde5744ecb074491b165ab9fb18e0165 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ed78e91842145e9bad2778034241a58 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09bb432718274710a17831a65848bee8 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1ff9ede53074dbcb5c77fe74f96e5d2 : SUCCESS in 13m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90c7d4c7e0ad4499a7fcdaec63156a7b : SUCCESS in 5m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/682c3061d3514c29a98b5ab8fbe1e482 : SUCCESS in 7m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/14908652d78746fea1b55d28cc2c5e4b : SUCCESS in 9m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1ee30a3b61b84285b3a15c0addbb7d17 : SUCCESS in 7m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a825e56a00fe4faa931bc57c88ea682d : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ecc5c71f3a7d40daafd03dd71a2d7dd3 : SUCCESS in 1h 56m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/89902f0d8a5a44aa97b29545cf069080 : SUCCESS in 1h 47m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/525c7a3ea88d45dc9721459d53051c09 : SUCCESS in 38m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/62e7ee16938d4acdacde2ec80dd64ace : SUCCESS in 33m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6897a7a8f4b04ceabeb6407798c2206a : SUCCESS in 35m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0928f70ac62240e5a54a83a95c5cd618 : SUCCESS in 57m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3eb8eef18dbc423cbf25b95b3735b751 : SUCCESS in 1h 55m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b32b5678ecd44e36abcc54305b40b6de : FAILURE in 32m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3126f6fc42ca4619896d5ec7a76a2762 : SUCCESS in 1h 22m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5e97467b6cdb4cf594e8ce369eda45ac : FAILURE in 1h 55m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3940e3e2e5bd4aa1a0805bfb2ecbb93a : FAILURE in 1h 43m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3a7d3ca598b04cd29745a0c744940d58 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b091de8a0083df44f4c0a307a9f8d01bf763538","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-25 08:36:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1d772e7127f4ea7b989e403f069ffc01f8dd80fe","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-25 11:55:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d225b4117fcf11c4c2f21f632955b54dc5c5f7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 11:56:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b7758822b63f75c5b3e72d9b12dd9fd6d3db9b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:17:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d034e09609464574aefa2e114d0b4238 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/332bc787d95b4bb0b9f05e60994f413b : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74964bb8934946fb82f7326e866ae16c : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e749020e4123492c881778235636fd8f : SUCCESS in 13m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30611d913bde438c8a04037399f09542 : SUCCESS in 5m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1d5167361a1a48fc94574b4437308ecf : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a1a411221ba43e2a39572a203789a33 : SUCCESS in 7m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e8e911dbc4b64f9c8d4dadf471f1299e : SUCCESS in 1h 33m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/269754b5e09d4a638f6a5810cde2ebd8 : SUCCESS in 1h 34m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aef9e3cc36e74c6f94167ffb1e0a90dc : SUCCESS in 35m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d1a9352bd579409a9b7c873eec4eefff : SUCCESS in 32m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0c3b36f2f9844056920f80ca10f63a97 : SUCCESS in 32m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a2ffc1585beb756ac13cf74d863610092e063b17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:17:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ad05c23c7f68c753ed4e571ecde71347b00da2af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:20:01.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa07685864074fbb899460ab310202b9 : SUCCESS in 1m 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23e346c9605f49cb81101ceb6de7c710 : FAILURE in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d4f2de8ba88a09a1139cf69be14acde06e765470","revisions":{"d4f2de8ba88a09a1139cf69be14acde06e765470":{"kind":"REWORK","_number":1,"created":"2022-04-20 14:41:56.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/36/838736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/838736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/838736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/838736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/838736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/838736/1"}}},"commit":{"parents":[{"commit":"5a54e1cb7bfd61cb66df6605df8c135c50cb5dbb","subject":"Merge \"Fix unplugging member ports\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a54e1cb7bfd61cb66df6605df8c135c50cb5dbb"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-04-13 11:52:53.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-20 14:41:44.000000000","tz":120},"subject":"Save the HAProxy state outside of its systemd unit","message":"Save the HAProxy state outside of its systemd unit\n\nBy default, SELinux prevents HAProxy context (haproxy_t) to execute\nshell context (shell_exec_t) for security reasons.\n\nThis prevents HAProxy to actually reload properly, since SELinux will\ndeny its call to a shell to save its state to a file.\n\nIn order to avoid opening a potential security hole in the load-balancer\nimage, the best way is to generate the state file before the actual\nreload.\n\nThere are more details about the SELinux denials in the associated Red\nHat Bugzilla.\n\nResolves: rhbz#2073491\nChange-Id: I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081\n(cherry picked from commit 21d74c373b92f8711ff91766c2e7c8bcb811bdda)\n(cherry picked from commit 0a062cd66416c6020d9718d87988727bb8a78627)\n(cherry picked from commit edcd6931fcc83dffc4d4a37c91b8db044ff3fdc4)\n(cherry picked from commit 97ac37bc6e6572ce7b368c974b928ce5e35700d1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4f2de8ba88a09a1139cf69be14acde06e765470"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4f2de8ba88a09a1139cf69be14acde06e765470"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
