)]}'
{"id":"openstack%2Foctavia~838737","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081","project":"openstack/octavia","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-04-25 15:17:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081","subject":"Save the HAProxy state outside of its systemd unit","status":"MERGED","created":"2022-04-20 14:42:10.000000000","updated":"2022-04-25 15:18:50.000000000","submitted":"2022-04-25 15:17:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838737","meta_rev_id":"c6a00b39d9afcc0da6c11042143f2933103e32e4","_number":838737,"virtual_id_number":838737,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-25 15:17:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-25 08:36:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-04-25 11:56:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-04-25 11:56:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"CC":[{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-20 14:42:10.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"CC"},{"updated":"2022-04-20 16:54:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-25 08:36:27.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-04-25 11:56:14.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"702de2b3bb43034e737b91bf5506706d7537455f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-20 14:42:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36da14a1098955dae447ab94677a565132bc2604","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 16:54:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78e397afba7847c1b4b2edb9213b031e : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/181862fd29a44d8499e0b6edf705942e : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3eb1cf2be80e40e2b48f8d4474f6b7f1 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/286bdf66670d4b0d95fc66d2613e2a85 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f162d19466a4b728b44b6714a107eed : SUCCESS in 11m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35a118b78f4244dfaf5824d9e6d6a3c7 : SUCCESS in 5m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/95f91670e9fb4b368362209d24d99d7f : SUCCESS in 7m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/de9dda83adc0486095be02ff14a86851 : SUCCESS in 9m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/706b2500e4634b2eba3a8dcec7ce38ed : SUCCESS in 7m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f39e67c87f314a51b605cad2634126df : SUCCESS in 1h 07m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f3e47d77546b49529f6cde204123689b : SUCCESS in 1h 19m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/406b03beaa3d4a0fbd8a7ef3d0a871e0 : SUCCESS in 36m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/115d2dba15a7466ab1f0e5f3b73ee5e6 : SUCCESS in 38m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/106c53af2b0f420eb439149d5ef5cebf : SUCCESS in 35m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/da796183c36044578dd1f095be18217e : FAILURE in 12m 21s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ad4fbc2e095840958e9da1e2daedfbad : SUCCESS in 1h 54m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6ed810ebb0ed44dea883807a6c1939ba : RETRY_LIMIT in 10m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dad630b15357427e9c5636202e56e352 : SUCCESS in 2h 01m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1089912e1c32881263e90edbbd2177dfe1c2f27","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-25 08:36:27.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1502bbc22d02bc82e88c6e79381f3b5ba355cc7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 10:31:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/746805f3f6a5494297e980b4d64c9cc8 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b377b7de784f442d9c3c24c1951980e0 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3aaf27bcba9848f7b6af6df7c193f29f : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b76b26cdae194a7882d0040ae2f6ba94 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77e8d3e6ed514eca80fbd2ca576e32a1 : SUCCESS in 12m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c26c18c41ef44ecaf5a7a0b18d32ad7 : SUCCESS in 5m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a8f037ddaa3842adbd383c72f5d47c77 : SUCCESS in 6m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/29cf1c427a0646cba4dd082da306881f : SUCCESS in 8m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2faf5c2ac6084e96a7a4bfbb1afc2ef8 : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4b61334a779a4552ac803c471cec82a1 : SUCCESS in 1h 05m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8f358e4bcef4279afbfdace384d72cf : FAILURE in 1h 49m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7407ea1f1a2e4ff69ad5aef308343104 : SUCCESS in 40m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1fb9d40968a345978b98d097f149e703 : SUCCESS in 40m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/65d3c3fc479f41968d957007727acb97 : SUCCESS in 38m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f68d997d8c46477f98db081bd33c6a1f : SUCCESS in 1h 49m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/089dbf25f7534d73bfd9668d87f2e66a : RETRY_LIMIT in 6m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/36e39f81789644f7964bcff84947b8f5 : SUCCESS in 1h 43m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af99252928f9d6b77b6f5a0824c2a0cceb9c1335","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-25 11:56:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9b1b3d3c252da681cd60d6d0b04e4ef7a7d0be39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 13:48:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37eb50b627a5494694dc9f7167185a05 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5443b73a2aa9434a92f351ea9247ad5c : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de998f30d4bf4e60853fc6b18542c60a : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29ec10215c064a98975b6c377d0fcf7b : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca7d794413c2407f90b53fe1128eb37d : SUCCESS in 13m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d08ece207be4a18ab5ea8ee5a352b90 : SUCCESS in 6m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7dca0921454d4a11be33a1481ddf01b8 : SUCCESS in 6m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f5cec00b493c43a495c35ddd869263bb : SUCCESS in 8m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a782751bb22f44a0a53d20e7fb6be526 : SUCCESS in 7m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/80031b0b87b340a0aa162b9e7c912b8f : SUCCESS in 1h 04m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3142b5e8fe3412bbb5206f095ede94c : SUCCESS in 1h 46m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dc1f672855e7414d9244e9c1954782a8 : SUCCESS in 34m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/111fdace7423410c8f1d30a7814926ef : SUCCESS in 33m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/87b4e65a64b548a193686c1bfcf1ff4d : SUCCESS in 33m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/da7994dd6e36491791ec4d4bd2af5846 : SUCCESS in 1h 42m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1afe547247534751a0b69d989e31593b : RETRY_LIMIT in 5m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/af55d5d746d5402eb1d4fff6c7ee9de6 : SUCCESS in 1h 48m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b247949a33a94345d224181ab63cd8aa37e1004","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 13:49:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9eb8033b7bf8e778b85a929002a5263fe06ceb06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 15:17:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eec21ca3aed4a7688cd91586c06d304 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/919d782a90c748bf9f0e9d2c3504742d : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b95cbef5460245db8dcb7ed0ac12d3ba : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9305b070a9d44208f5482af641a57b0 : SUCCESS in 14m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fe5004ed0cb4891a2b32db304f4a0ea : SUCCESS in 5m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2eff2dc0bb55431d92a112c9d5fcadff : SUCCESS in 7m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e832b6840a1a4a8ca9903d023e5f627c : SUCCESS in 1h 13m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0ea3daeadba543048d2e83f8da9ccf29 : SUCCESS in 1h 26m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5751f53a4ac74b90b3833eb6a1a60eac : SUCCESS in 39m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6a6307b1aff14a898742fade7e427434 : SUCCESS in 37m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bb7c62e5f8a847148f89ba2b9ec9b257 : SUCCESS in 36m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e2d2428e8392dc4fe60f6c731363153b9f5da8c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 15:17:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c6a00b39d9afcc0da6c11042143f2933103e32e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 15:18:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90656064ce3c41fb8d186634e40e3aa5 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5372a01088764dd8916931f1fa43a47d : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"86672909d435ddb1941e5b4fb0ef92d806c4802d","revisions":{"86672909d435ddb1941e5b4fb0ef92d806c4802d":{"kind":"REWORK","_number":1,"created":"2022-04-20 14:42:10.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/37/838737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/838737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/838737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/838737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/838737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/838737/1"}}},"commit":{"parents":[{"commit":"3d56e7ad20ec3bf3280ca10f7e3d3754ea197b71","subject":"Merge \"Fix unplugging member ports\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3d56e7ad20ec3bf3280ca10f7e3d3754ea197b71"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-04-13 11:52:53.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-20 14:41:57.000000000","tz":120},"subject":"Save the HAProxy state outside of its systemd unit","message":"Save the HAProxy state outside of its systemd unit\n\nBy default, SELinux prevents HAProxy context (haproxy_t) to execute\nshell context (shell_exec_t) for security reasons.\n\nThis prevents HAProxy to actually reload properly, since SELinux will\ndeny its call to a shell to save its state to a file.\n\nIn order to avoid opening a potential security hole in the load-balancer\nimage, the best way is to generate the state file before the actual\nreload.\n\nThere are more details about the SELinux denials in the associated Red\nHat Bugzilla.\n\nResolves: rhbz#2073491\nChange-Id: I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081\n(cherry picked from commit 21d74c373b92f8711ff91766c2e7c8bcb811bdda)\n(cherry picked from commit 0a062cd66416c6020d9718d87988727bb8a78627)\n(cherry picked from commit edcd6931fcc83dffc4d4a37c91b8db044ff3fdc4)\n(cherry picked from commit 97ac37bc6e6572ce7b368c974b928ce5e35700d1)\n(cherry picked from commit d4f2de8ba88a09a1139cf69be14acde06e765470)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/86672909d435ddb1941e5b4fb0ef92d806c4802d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/86672909d435ddb1941e5b4fb0ef92d806c4802d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
