)]}'
{"id":"openstack%2Foctavia~838741","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081","project":"openstack/octavia","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-04-25 14:17:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081","subject":"Save the HAProxy state outside of its systemd unit","status":"MERGED","created":"2022-04-20 14:52:59.000000000","updated":"2022-04-25 14:20:08.000000000","submitted":"2022-04-25 14:17:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838741","meta_rev_id":"7ca12bfc371f7d4fba3a134df7c93a3af5b9826b","_number":838741,"virtual_id_number":838741,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-25 14:17:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-25 08:35:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-25 11:56:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-25 11:56:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-20 14:52:59.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"CC"},{"updated":"2022-04-20 15:33:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-25 08:35:52.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-04-25 11:56:39.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"d45cf017139dea2a2dacc325ed359d844a33f4d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-20 14:52:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb1beddd2dd63fa6a478cb40882b7766006ccde5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 15:33:05.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de75675205794281bb4df3063cacc45f : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad52ccb6e05b421d921c748329763d88 : FAILURE in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a41fc3333b7e4ddd90c1dc689726c3ec : FAILURE in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/930517dfaba14399a4537ad27c62fa4e : SUCCESS in 6m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f99c5baf85264bbe978f33c36198f465 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c8b2fbd687f46d9879d6408b3b64f31 : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/751b475bcae947e0ac4cce18046fdb45 : SUCCESS in 6m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/205e1489ff74425eb3ef4d0233ddfa24 : SUCCESS in 7m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f44e3666cf30426b88dd6825e90c0b6a : SUCCESS in 9m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f67373710bb0422ca644a26993c478b3 : SUCCESS in 8m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d63cb2ab8b584f8eaa926f9c269f8d4f : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e071b74eab2f4b80939e43c73f71156f : FAILURE in 15m 33s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e1be573358904073ad3e0d4fb356bd18 : FAILURE in 13m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/662e066314fd466ba2edb232fd0e71a8 : FAILURE in 12m 59s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f2c04233bbfa4e629222dd6c73fc04c7 : FAILURE in 13m 14s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/97582a1e83764828932f294d0f07dbab : FAILURE in 18m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6a089a4ad0aa4a0395094360020db9da : RETRY_LIMIT in 7m 27s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/922b041b75174e759ba751df3ef6406a : RETRY_LIMIT in 8m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4208c29cd58f499ab644767a7e651952 : FAILURE in 13m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/704de477c5804f2488b42aa12cff2783 : FAILURE in 14m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d02c1c73d54f49f8b5b2efb6b4777761 : FAILURE in 13m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6b7fa56b2a85431bbdfe9d1b865161e4 : FAILURE in 14m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/381e66ec3b0a42b3abb8ea324e1f9878 : FAILURE in 18m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd37283e3f180542fa56508e5fe8f354b5be636c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-21 09:29:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2aa9986c6e4dbc4390137714f061109f961bd6ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-21 09:36:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a36dd733277e1afff4d77eec12a49c92ef3021a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 11:55:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2f5fed02a7b472aa5cb7844f61d9288 : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c052b7f42c1e481abe1057156041ee3f : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d940fa706f54c2ebc380cd507189561 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97aa3f07cf2f45de8c22f72cd6d8caa4 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4a57c8f335843d1addc81fcbd8436c6 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d07bffc90f3340cf809d4b24cbe26bc7 : SUCCESS in 10m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b7a163080534b95a1e8693cfbb3a80e : SUCCESS in 4m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b0fed7b260164801abb3cbab5fdccdf4 : SUCCESS in 5m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9d9898caed1c47139167c56f8609653a : SUCCESS in 7m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3d12f94af71b4075afef18726d0eff44 : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/864dd7a9e07c440ba2ab936945ba6ca9 : SUCCESS in 6m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cbfb4c1295c6481abe8be393d378597e : SUCCESS in 1h 05m 53s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e4f4e052ece34e1ab4a762cbf75e4c6f : SUCCESS in 1h 10m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/00cc5a1bccc24485b9ff93075465b314 : SUCCESS in 1h 34m 53s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/756cf192f73347839ca4b17074cde184 : SUCCESS in 2h 13m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5cb086fd74f04567b567bba9a390dd73 : FAILURE in 16m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e096c075c3db41d883dd718c5338d12d : RETRY_LIMIT in 6m 36s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/067483d04e8a49d2a107ad4d4ce588f6 : RETRY_LIMIT in 6m 35s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1d3963c2cf5e476890b8615dee735ac2 : SUCCESS in 41m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/84b51959fa3f4dcbb563309e3c9ce938 : SUCCESS in 1h 54m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/24bdd4b3af8d4bc2b8fb21804abaaa0b : FAILURE in 35m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1fb5773e51134852a34bd3026ccabdaf : FAILURE in 34m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fa3d675e608745188f21c3fa52077951 : FAILURE in 14m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"54fd796ab19a90067ce00dee2abb62dc87dbba4a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-25 08:35:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0950a2794af9aacca488d750ed34b7ab04377e27","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-25 11:56:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"aa57b93a38bf48559bbf6b44747490d6af52e999","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 11:57:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bc9b658a8cacb36223e63981dbd6a00ac98635c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:17:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fa9bb8233c54835a3e218a679e8e9ca : SUCCESS in 6m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a14edd6bf0694fadad43a54780a5ff99 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0de6c80446984ac59e02bea3a27d0672 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4ab60d13ae24568be69c6601f0f8914 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e47e329b691240169d3dc337128c8e29 : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5aabd2b5b3ec4af9a6d1fc7d0d3ff237 : SUCCESS in 5m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b0b714909f19431fa7e484306d41f82a : SUCCESS in 6m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fb153f7294904dd196298f10ab24c69e : SUCCESS in 6m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b6e2b4ed061e4b90b2eb29e384932acd : SUCCESS in 1h 03m 59s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9878fa30c15f44b08e5cf11cfc87d2a1 : SUCCESS in 1h 08m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fef68765024d4ae7ac9a62762faeb2ab : SUCCESS in 1h 36m 43s","accounts_in_message":[],"_revision_number":3},{"id":"17c7fd5c8c8d8e5cc76585cb59ffcb6a6e795905","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:17:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7ca12bfc371f7d4fba3a134df7c93a3af5b9826b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:20:08.000000000","message":"Patch Set 3:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e58e25dc7e345deb34dc5db28cb0ece : SUCCESS in 1m 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe3e82a0a98947e99726c2aedc0fd2d0 : FAILURE in 1m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9922248a896c9ac836055b7e4fe421f796f835b8","revisions":{"814524eb110df749159d21c81d2a03064d58b222":{"kind":"REWORK","_number":1,"created":"2022-04-20 14:52:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/41/838741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/838741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/838741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/838741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/838741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/838741/1"}}},"commit":{"parents":[{"commit":"bbf18f97ea1e18faa1bb9faf98fd03706e8344ff","subject":"Merge \"Fix unplugging member ports\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bbf18f97ea1e18faa1bb9faf98fd03706e8344ff"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-04-13 11:52:53.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-20 14:51:50.000000000","tz":120},"subject":"Save the HAProxy state outside of its systemd unit","message":"Save the HAProxy state outside of its systemd unit\n\nBy default, SELinux prevents HAProxy context (haproxy_t) to execute\nshell context (shell_exec_t) for security reasons.\n\nThis prevents HAProxy to actually reload properly, since SELinux will\ndeny its call to a shell to save its state to a file.\n\nIn order to avoid opening a potential security hole in the load-balancer\nimage, the best way is to generate the state file before the actual\nreload.\n\nThere are more details about the SELinux denials in the associated Red\nHat Bugzilla.\n\n Conflicts:\n\toctavia/amphorae/backends/utils/haproxy_query.py\n\nResolves: rhbz#2073491\nChange-Id: I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081\n(cherry picked from commit 21d74c373b92f8711ff91766c2e7c8bcb811bdda)\n(cherry picked from commit 0a062cd66416c6020d9718d87988727bb8a78627)\n(cherry picked from commit edcd6931fcc83dffc4d4a37c91b8db044ff3fdc4)\n(cherry picked from commit 97ac37bc6e6572ce7b368c974b928ce5e35700d1)\n(cherry picked from commit d4f2de8ba88a09a1139cf69be14acde06e765470)\n(cherry picked from commit 86672909d435ddb1941e5b4fb0ef92d806c4802d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/814524eb110df749159d21c81d2a03064d58b222"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/814524eb110df749159d21c81d2a03064d58b222"}]},"branch":"refs/heads/stable/train"},"de2dfcaeafaed5a906e9cdc01614daf5d2286b21":{"kind":"REWORK","_number":2,"created":"2022-04-21 09:29:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/41/838741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/838741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/838741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/838741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/838741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/838741/2"}}},"commit":{"parents":[{"commit":"bbf18f97ea1e18faa1bb9faf98fd03706e8344ff","subject":"Merge \"Fix unplugging member ports\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bbf18f97ea1e18faa1bb9faf98fd03706e8344ff"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-04-13 11:52:53.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-21 09:27:12.000000000","tz":120},"subject":"Save the HAProxy state outside of its systemd unit","message":"Save the HAProxy state outside of its systemd unit\n\nBy default, SELinux prevents HAProxy context (haproxy_t) to execute\nshell context (shell_exec_t) for security reasons.\n\nThis prevents HAProxy to actually reload properly, since SELinux will\ndeny its call to a shell to save its state to a file.\n\nIn order to avoid opening a potential security hole in the load-balancer\nimage, the best way is to generate the state file before the actual\nreload.\n\nThere are more details about the SELinux denials in the associated Red\nHat Bugzilla.\n\n Conflicts:\n\toctavia/amphorae/backends/utils/haproxy_query.py\n\n Fixed in this backport:\n   - A compatibility issue with Python 2.7\n     (open() didn\u0027t have an encoding kwarg)\n   - save_state() tried to use Exception().output, which in general\n     causes an AttributeError\n\nResolves: rhbz#2073491\nChange-Id: I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081\n(cherry picked from commit 21d74c373b92f8711ff91766c2e7c8bcb811bdda)\n(cherry picked from commit 0a062cd66416c6020d9718d87988727bb8a78627)\n(cherry picked from commit edcd6931fcc83dffc4d4a37c91b8db044ff3fdc4)\n(cherry picked from commit 97ac37bc6e6572ce7b368c974b928ce5e35700d1)\n(cherry picked from commit d4f2de8ba88a09a1139cf69be14acde06e765470)\n(cherry picked from commit 86672909d435ddb1941e5b4fb0ef92d806c4802d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de2dfcaeafaed5a906e9cdc01614daf5d2286b21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de2dfcaeafaed5a906e9cdc01614daf5d2286b21"}]},"branch":"refs/heads/stable/train"},"9922248a896c9ac836055b7e4fe421f796f835b8":{"kind":"REWORK","_number":3,"created":"2022-04-21 09:36:51.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/41/838741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/838741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/838741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/838741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/838741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/838741/3"}}},"commit":{"parents":[{"commit":"bbf18f97ea1e18faa1bb9faf98fd03706e8344ff","subject":"Merge \"Fix unplugging member ports\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bbf18f97ea1e18faa1bb9faf98fd03706e8344ff"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-04-13 11:52:53.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-21 09:36:38.000000000","tz":120},"subject":"Save the HAProxy state outside of its systemd unit","message":"Save the HAProxy state outside of its systemd unit\n\nBy default, SELinux prevents HAProxy context (haproxy_t) to execute\nshell context (shell_exec_t) for security reasons.\n\nThis prevents HAProxy to actually reload properly, since SELinux will\ndeny its call to a shell to save its state to a file.\n\nIn order to avoid opening a potential security hole in the load-balancer\nimage, the best way is to generate the state file before the actual\nreload.\n\nThere are more details about the SELinux denials in the associated Red\nHat Bugzilla.\n\n Conflicts:\n\toctavia/amphorae/backends/utils/haproxy_query.py\n\n Fixed in this backport:\n   - A compatibility issue with Python 2.7\n     (open() didn\u0027t have an encoding kwarg)\n   - save_state() tried to use Exception().output, which in general\n     causes an AttributeError\n\nResolves: rhbz#2073491\nChange-Id: I6b9a5e1e3bafe77ad9f9506b8c0995d8c2a00081\n(cherry picked from commit 21d74c373b92f8711ff91766c2e7c8bcb811bdda)\n(cherry picked from commit 0a062cd66416c6020d9718d87988727bb8a78627)\n(cherry picked from commit edcd6931fcc83dffc4d4a37c91b8db044ff3fdc4)\n(cherry picked from commit 97ac37bc6e6572ce7b368c974b928ce5e35700d1)\n(cherry picked from commit d4f2de8ba88a09a1139cf69be14acde06e765470)\n(cherry picked from commit 86672909d435ddb1941e5b4fb0ef92d806c4802d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9922248a896c9ac836055b7e4fe421f796f835b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9922248a896c9ac836055b7e4fe421f796f835b8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
