)]}'
{"id":"openstack%2Foctavia~838848","triplet_id":"openstack%2Foctavia~master~I36acdee1f8782b17c234f6b250facc5c8d0aaf87","project":"openstack/octavia","branch":"master","topic":"rhbz2073491","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-04-21 12:21:14.000000000","reason":"Gregory Thiemonge replied on the change"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-04-22 10:46:11.000000000","reason":"Change was submitted"},"28223":{"account":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"last_update":"2022-04-22 10:46:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I36acdee1f8782b17c234f6b250facc5c8d0aaf87","subject":"Fix AttributeError in exception handler","status":"MERGED","created":"2022-04-21 09:44:21.000000000","updated":"2022-04-22 10:48:10.000000000","submitted":"2022-04-22 10:46:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838848-rhbz2073491","meta_rev_id":"e28aac452d36dbfd0677f68cd865a1b4542090b3","_number":838848,"virtual_id_number":838848,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-22 10:46:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2022-04-21 12:21:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-04-21 17:13:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-04-21 17:13:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2022-04-21 12:21:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-21 09:45:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-21 09:45:19.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2022-04-21 09:45:19.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-04-21 12:28:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-21 17:13:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"0cc707ca37962cd7204817c58b66514adf9d2f27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-21 09:44:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef72b703d8a7a813d0b05061e8e7ff555a863c1c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-21 09:45:19.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"df91562ad96e457914784dd052215c14b2229371","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-21 12:21:14.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"12fe8de296560f696ce346b835f3dc8216639d86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 12:28:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a4415b15d0449aaa68038ba892500fb : SUCCESS in 11m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7154cf9a06454850888f2547f26667a1 : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71573064725b4b6088b1017be33a5f9a : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a9da845a26841a1845fb91d22a13f57 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e58d11d4556941649358c9062c6c7a1e : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1510652432b4893b62b283eebf21eb7 : SUCCESS in 9m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f45889013e5d46ed8edb1bc969dba01e : SUCCESS in 14m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1da292e747734f39a27daf15e1582fea : SUCCESS in 6m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1495616b52df43b1b65c2319574ee6b2 : SUCCESS in 10m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6abd4ebf62f541c0b57937488859002e : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4698bff638a24372817d621af0260f3a : SUCCESS in 9m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d5149ce9e6c34eacb763d12ec53a5b72 : SUCCESS in 2h 37m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f0f6760f3434a7194b03cfdaaaaa79c : FAILURE in 2h 06m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5dcc5ab37bf345bf86c148b2a64f8444 : SUCCESS in 38m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c6774b1e933f43e195f79bd9adeceee3 : SUCCESS in 1h 07m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6b99892cdec3435991df151ef1a5b77a : SUCCESS in 58m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/346b6f6c55424c719aca4d67a7f86da8 : SUCCESS in 32m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b743282539a6413580192ea97ce349f2 : SUCCESS in 1h 54m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/27e26de9d2b54cb5b5f5a0ca9cf1ec39 : SUCCESS in 1h 41m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d46d8a66584c4d88bed21b1f712dc07b : SUCCESS in 2h 12m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/921558998b1c48ccbf290eb71b3fb0d9 : SUCCESS in 1h 43m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8c3f85e5e28345aab889b83367f52714 : SUCCESS in 1h 25m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe73c543bf1d4e63879f5844c394854a : SUCCESS in 2h 03m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1e0055904d8366b41c6c089dd8458c1943c0dc4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-21 13:59:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51009097d372f0de76f3ca0e85bfc9ba529cb7ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 16:12:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba7c97204ddd4b5b91e3ebb58d0cd060 : SUCCESS in 11m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01862a11cdff4678a43798bfe4341587 : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa943c53d8d44d908f017625a910c8c3 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27043b6f38ea4dc28f63a6aad765d1e2 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd7418baf13c4c0194f187cce600de6c : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db15b3d4650d4ca7985f761c61045573 : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d387274e2234ae396de5705170366ed : SUCCESS in 11m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f3e44b94654a4773af3cad075b691878 : SUCCESS in 6m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7d0fcd71cef1426f9fb3581bb85854c2 : SUCCESS in 9m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/23c140c1c0ed4041938ce4250acabf80 : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/374830e3aee44c09af6e8684a6541021 : SUCCESS in 8m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/37d814f200e44cc8b17d4df92278f496 : SUCCESS in 1h 56m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bd7aa8d5f05944438d34f68c190b0e80 : SUCCESS in 1h 51m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6f460d61718a46d080505c8fe7946aa6 : SUCCESS in 40m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0462b0e831a941c990f8ef75c21c31ba : SUCCESS in 1h 11m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/175568d31c374042acb09ba007c6ae35 : SUCCESS in 1h 03m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bff03b0f849741a98ebe2ec093467c8d : SUCCESS in 42m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3dce127126e94a69ba355ee7da17e756 : SUCCESS in 1h 24m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bff330806f8f4e8d84666bd8b37cc5ac : SUCCESS in 2h 05m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8889b34040f94ae4995059987e8a94a8 : SUCCESS in 1h 30m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4e6fb0ecc0b5418e9232d591b6ef0b3c : SUCCESS in 1h 34m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9223573e30474e9698c09c4425694da7 : SUCCESS in 1h 18m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd174b619bed4124840c32a9a059cc50 : SUCCESS in 1h 46m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"048ddc2fe671090ba1dbd552c54eb3e66bae187b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-21 17:13:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ca9f66befec1d8af498d1ab564f2cd7c2cfee9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 17:14:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02de4237ce6e95c82a84af8d4a80d79b53ffb6c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 18:55:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3b09cd9b10b462ca4f96ce504daa833 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/544be8e5da4c417eb7568f794e25e0d7 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3712d904f8ef49adad8cc52345924778 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b801daf0a0ef4a2d83c41bcc6e1f5d89 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4941749b5ae46f79cd70dcbaf887178 : SUCCESS in 11m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/668d1695762b4fc0a0fd6194238f0e4f : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/67c325e2f9de43bdaab87b0501e6c181 : SUCCESS in 8m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/99a45b3cdd844f69bfbaf7995ca2085a : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab49b751682b4c0fae1bcb91e6824654 : FAILURE in 1h 38m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0bc80080fa884444885a9fbd6df23fc0 : SUCCESS in 40m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b5e827e7a5924ab39a379a621c1e9423 : SUCCESS in 1h 08m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7f0bf0a0c9d3e674cf5fe81285f80606fb101534","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-22 05:47:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f81240fa54d07a4aa5d7fb4eafcfd9af3dca8dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 08:22:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03782993bf5d458fa75dda1e91cee434 : SUCCESS in 9m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d84fe436a8084987a4aa309f45509a91 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad4292aa50484dcf973914a116e57446 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df230d3364b54c7c9d52c94163391d48 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c991ed517264160baceb172e4ac8874 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10acade88aef45d996fdaf982d6e9027 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f7e25a93c4e4b9b9fcbfe6425d82396 : SUCCESS in 9m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b9ae3e0567ae4920abdf36c7b74d9d62 : SUCCESS in 5m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/67bdfd34dee441d496edf518cd110283 : SUCCESS in 8m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4938683c01404cb590af503afaa54a51 : SUCCESS in 4m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bde96ac6a17546b8b9bd42455a53c19a : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4f3b04f1826f44adb601b6617896b23f : SUCCESS in 1h 52m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74042d4a83dc421fa88aa38c35861112 : SUCCESS in 1h 31m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8d60c5c173954a14bacf4325832d7ae0 : SUCCESS in 34m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/98728f8fb9364235a9fa5273d288e4f1 : SUCCESS in 50m 48s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5a20433eda824b2f945f53f9e9d8f21f : SUCCESS in 50m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d8e693fafbd44e5bbae8dcba037683e : SUCCESS in 33m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4c7f1add0e314f28ab7907296089b631 : SUCCESS in 1h 32m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ea065a902a1a423bbecc2892cbdb434b : SUCCESS in 2h 18m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dc7dc7aa0b454ef0b6919e77825b1f1e : SUCCESS in 1h 31m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a4b647c83ac247cdbf7b77405cc961c8 : SUCCESS in 1h 32m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/55129a5254a94d329f937f7b27b40c55 : SUCCESS in 1h 31m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/665ad42a4b1543efad25ee126f0983a5 : SUCCESS in 2h 08m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"591356b043b9403b435863acc2facd83ca7415ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 08:23:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"326db632f471654081a4bacfb324141f3084a251","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:46:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36ee1c451b724324be0f24d10fc9c065 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd8ec899e4f8475d836c34c2bceeb65f : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92f8d948332d419ba3339377146cfdfb : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3a4c05e221e40ec9a8013de4e09ac99 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a56c122d12fb4a64a4b470a26bd78086 : SUCCESS in 11m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8d8a11f1b0094dc292aaa29ccfb3bd8d : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d7354787e2714225b1f783c946b430d1 : SUCCESS in 8m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/24b13a149fb74cdea6f6beb44578dcf4 : SUCCESS in 2h 22m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd229ccc312045208975fa99ca819f74 : SUCCESS in 1h 21m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e8b52cfe18a0493cbe558f687034d31c : SUCCESS in 34m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3918b348fb474fc9ba09746eff89c973 : SUCCESS in 53m 02s","accounts_in_message":[],"_revision_number":1},{"id":"78446b3ababea11df270774fab248f3f58128e5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:46:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e28aac452d36dbfd0677f68cd865a1b4542090b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:48:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09659f9f11f940beb2252410eca508d2 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"384b83d88b436c8e657b09f97b12e42d7c9c339b","revisions":{"384b83d88b436c8e657b09f97b12e42d7c9c339b":{"kind":"REWORK","_number":1,"created":"2022-04-21 09:44:21.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/48/838848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/838848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/838848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/838848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/838848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/838848/1"}}},"commit":{"parents":[{"commit":"21d74c373b92f8711ff91766c2e7c8bcb811bdda","subject":"Save the HAProxy state outside of its systemd unit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21d74c373b92f8711ff91766c2e7c8bcb811bdda"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-21 09:40:57.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-21 09:40:57.000000000","tz":120},"subject":"Fix AttributeError in exception handler","message":"Fix AttributeError in exception handler\n\nThe exception handler of save_state() caused an AttributeError\nbecause it tried to get the output attribute of the Exception,\nwhich does not exist in general. Added a test to cover this\nscenario.\n\nChange-Id: I36acdee1f8782b17c234f6b250facc5c8d0aaf87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/384b83d88b436c8e657b09f97b12e42d7c9c339b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/384b83d88b436c8e657b09f97b12e42d7c9c339b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
