)]}'
{"id":"openstack%2Foctavia~839043","triplet_id":"openstack%2Foctavia~stable%2Fxena~I36acdee1f8782b17c234f6b250facc5c8d0aaf87","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-04-27 20:06:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I36acdee1f8782b17c234f6b250facc5c8d0aaf87","subject":"Fix AttributeError in exception handler","status":"MERGED","created":"2022-04-22 10:53:53.000000000","updated":"2022-04-27 20:08:10.000000000","submitted":"2022-04-27 20:06:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839043","meta_rev_id":"e8bf237fa3efe12234691c9f5fcc95f2a9dbb0b0","_number":839043,"virtual_id_number":839043,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-27 20:06:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-25 08:38:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-04-22 15:05:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-04-27 12:48:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-22 13:10:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-22 15:05:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-04-25 08:38:23.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-04-27 12:48:04.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"18a3c188dfcadd43683b2f6545b0c36af1016905","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-22 10:53:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d107f93e254a97e10dfa537da265fa78e0309e96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 13:10:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4225e3bf7e8a4931971ee48ac395b54a : SUCCESS in 9m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d0f984e349e42319b33931515a8a826 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd0cc6934ab34e0bb1fa8f951456c218 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8c6d0c436fa4178bf00fb5178ba7b35 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8052fae692bb43368a6c2376a62d2cb2 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/656d6015aca640b4b0b078997b8cb347 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5f6e19b989c48cfba9439871b73dc5f : SUCCESS in 12m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bd74203bf21240debe77d106d7d74038 : SUCCESS in 5m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/eb328a2edc4046268a3dd1b078ae78ed : SUCCESS in 9m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6534bc37b41140898dd0acef5054062d : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4878abcaf07b437d920783623ef10fe4 : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ec13d385bc294e268c4e69ebaa290e01 : SUCCESS in 2h 05m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2af84c6e4a0b49d1b5154a35f0143d0a : FAILURE in 1h 51m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8c12ba3a78ec4755aea498be8c235ba7 : SUCCESS in 37m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4615996c542a4d4d93fc4366a66c3271 : SUCCESS in 55m 02s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/275e522b276e4d0a904a2c9e833643c1 : SUCCESS in 1h 50m 22s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/82e92107516c4ec5aafa1578925b04b8 : SUCCESS in 45m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74044fd140ce42d389e2e196d33239ac : SUCCESS in 36m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/196e1cd027a1435cbcae0f29fcf60e97 : SUCCESS in 1h 43m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/508a97be1c69450b91b630dd12b69552 : SUCCESS in 1h 48m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/70eccac016c94c4baf06812501c632a2 : SUCCESS in 2h 11m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/29627c21af0249399b544bc2d3f1c74f : SUCCESS in 1h 45m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2d3ccd12c37941dd98c9a2f887177242 : SUCCESS in 1h 37m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/51ac8a46bbcd4ecaaf547e1529cdf9a7 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa8f9e608b31d28e9879de5de0ee569d910a5463","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-22 15:05:57.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f830bcbc8c04fc651cb400bf3db9cf78651dad11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 17:28:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92bd9585e54540daa4ab26b3efa7cff5 : SUCCESS in 11m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12ef32b1c3fd409d9681a673670e23de : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13189e51173543908c8dd73e9473aa65 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32bb826442b5475fabd736019233ed35 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/487a9bfe39114df3bdd7926f319210cf : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e90f4737c159498b9ec4766186d2cf1f : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2747dcadc2234cdca24cbe05b05c04a5 : SUCCESS in 12m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f88edc83e9dc4949823af6efb1d1961a : SUCCESS in 5m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/564bf208177541d6b544b0122035aa5b : SUCCESS in 10m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/37278a6dc8614e54b5cfa51a1f7a6fe4 : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fedbb37594b548ebb990004ef6c51cc3 : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f8f76c16acb04a4aab57b252525dfe44 : SUCCESS in 2h 16m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a1dd400185104caab6a5a081ffaaef8d : SUCCESS in 1h 28m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/53d9e0d5a44943f5920182c36c86b22b : SUCCESS in 38m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b275cb5efc642e2b1cf8462ba512b71 : SUCCESS in 52m 11s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/441d700c94ad4cb0a9280548e1ecf4b1 : SUCCESS in 1h 58m 27s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e75148810a1d4e3b99af77bdbbe501cc : SUCCESS in 48m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17a97326ddb94f02b2ad1274ff5a23a9 : SUCCESS in 32m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5d68e209c88a465eb11530eb9462a49b : SUCCESS in 1h 42m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7087bfcf1ca3438da544e4c0cff7b593 : SUCCESS in 1h 44m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c21207ceb3ec473dbf71ee4452469dce : SUCCESS in 1h 50m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/801d415d3d554832a10176b7009c052a : SUCCESS in 1h 38m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3a5a3c7a46eb4f17aed9dadd4b9e15c4 : SUCCESS in 1h 38m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e3e7712a20214c3eba4fbab29162835b : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"55c52fbb00e0bc40867199693c4cc45f421c0e93","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-25 08:38:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7766e8abe716514e96f41951563bb5c92d1037f4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 12:48:04.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b854dfcc52e644adbc65050e712eac3397263528","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 12:49:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5963572463092ae079d1bf02b1be9f480919c8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:00:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aee87024303e4812b11024b5cda38811 : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89408601f1b24d12832795515697a311 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/384922bc44914bb78fe101a3f9cc3dbd : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc42b530fac04cccb1d7b83301286956 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fcb891c3f6d4de488705e2314a3180c : SUCCESS in 11m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9711d18f4d4241f89b72a96094724d38 : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f7df2e0725144b1f8bd6742ddb05334f : SUCCESS in 9m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b6d5b8c7fd114bf0bee98ea650390940 : SUCCESS in 2h 03m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/256e085661fc47ac89ba7c48323231df : FAILURE in 2h 06m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0bac40c3a17747dab338f9f25dd74034 : SUCCESS in 40m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2f0427ab19404506a002803756ca9a91 : SUCCESS in 47m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e52763f622c9a5e66f94c013b4983525ac1dedac","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 15:03:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"560bf5d8b8e2cbcda1729f87378c7d300375eb81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 18:01:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7757cc175faa44c0965394e8001327c6 : SUCCESS in 10m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96c4724916244591afe680e86c6474fd : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f26cf07170241ada4eb70212b16dd3e : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad5db88df2144e518fcb36278f4205a5 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee20516698e64090bef1d1023188482d : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8f6788cea3b4679a46f8780dcb39a3e : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/753d788595214b2788658e15777ba600 : SUCCESS in 12m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/251894d961104869ad12194c76ece6d0 : SUCCESS in 5m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e5736cd0f44942aa8739aa65f73314b8 : SUCCESS in 10m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9d858fd9f00d4273892c560725926403 : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/99cf3c332cad4eea83ed214bd1a0d092 : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a9b8f21c91ff402e9503eb8c7d2abfa4 : SUCCESS in 1h 52m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d190c3d1b614486b12dd06cef262a69 : SUCCESS in 1h 27m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/89d3b4e3e9644de5865b5905e0f78fc6 : SUCCESS in 37m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f20e3d67dad34eacb0241168f0b64efc : SUCCESS in 1h 06m 26s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/311bebebe95c415ca7730cbc45700549 : SUCCESS in 2h 06m 00s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b914f8cc98254813ae1bf453cfb87446 : SUCCESS in 49m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a66965dfce924315b850799855a3b0ac : SUCCESS in 37m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/accca700b0384249b3c1e91620f1326c : SUCCESS in 1h 42m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0fd4a84c077a4b81abed3b458d8da3c2 : FAILURE in 2h 44m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eb2f6316188245af889b7c79f26aacd5 : SUCCESS in 1h 47m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5595bddf20654fd183ae09f5c1fa9c28 : SUCCESS in 1h 36m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e4bb56a0a30340bd9e1d87f73ef58cd0 : FAILURE in 1h 51m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bc16c542c7924b10bc1ed106c4f9d494 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bad86998864181ab3b048f5fb7fd4276ab738d44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 18:02:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c55338d064cec50dbb5d2397019e700d4283b416","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 20:06:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc97ea8d8cd84ffe9b338df543a2c2e3 : SUCCESS in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0b741157d49423c8d917d2b2f6a197c : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c2b4b7dae794f04bc15385dd0c41779 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3df8e5bfc7c4301bb26a34bdf1f662b : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e1212089f9439ab2c38a50294642fd : SUCCESS in 13m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f5e4e25cce044faa863534d85f2e5081 : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e6d3dfec497c4621ba6510ff687d77b2 : SUCCESS in 8m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/46ca0ea39bb74e5cbd31b8307610963e : SUCCESS in 1h 59m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9c4ae2a6714147a6ae2f54a7ebe048ea : SUCCESS in 1h 37m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9845b37d07264ce59facaae2205e0af7 : SUCCESS in 40m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/99817c2c740343239508773db0f6f0fa : SUCCESS in 1h 32m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f2a2e53047de14a68f5b25c306c9ca98deb7bec6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 20:06:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e8bf237fa3efe12234691c9f5fcc95f2a9dbb0b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 20:08:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c603963993c74351928883b737179347 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7adeb21c42fdb6431718fcfcd9ad53b8ae077aa5","revisions":{"7adeb21c42fdb6431718fcfcd9ad53b8ae077aa5":{"kind":"REWORK","_number":1,"created":"2022-04-22 10:53:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/43/839043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/839043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/839043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/839043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/839043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/839043/1"}}},"commit":{"parents":[{"commit":"edcd6931fcc83dffc4d4a37c91b8db044ff3fdc4","subject":"Save the HAProxy state outside of its systemd unit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/edcd6931fcc83dffc4d4a37c91b8db044ff3fdc4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-21 09:40:57.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-22 10:53:37.000000000","tz":120},"subject":"Fix AttributeError in exception handler","message":"Fix AttributeError in exception handler\n\nThe exception handler of save_state() caused an AttributeError\nbecause it tried to get the output attribute of the Exception,\nwhich does not exist in general. Added a test to cover this\nscenario.\n\nChange-Id: I36acdee1f8782b17c234f6b250facc5c8d0aaf87\n(cherry picked from commit 384b83d88b436c8e657b09f97b12e42d7c9c339b)\n(cherry picked from commit 7aaac224cb4296a4802e4c21c76ba9d010335333)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7adeb21c42fdb6431718fcfcd9ad53b8ae077aa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7adeb21c42fdb6431718fcfcd9ad53b8ae077aa5"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
