)]}'
{"id":"openstack%2Foctavia~839044","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I36acdee1f8782b17c234f6b250facc5c8d0aaf87","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-04-27 14:55:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I36acdee1f8782b17c234f6b250facc5c8d0aaf87","subject":"Fix AttributeError in exception handler","status":"MERGED","created":"2022-04-22 10:56:05.000000000","updated":"2022-04-27 14:57:52.000000000","submitted":"2022-04-27 14:55:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839044","meta_rev_id":"7d3cb0e040786fc7b4e33e0e2bb0920c7537430a","_number":839044,"virtual_id_number":839044,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-27 14:55:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-25 08:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-04-22 15:05:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-04-27 12:47:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-22 13:22:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-22 15:05:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-04-25 08:37:59.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-04-27 12:47:55.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"fe15c8967f1ae365197180d0c8a0fdc5a84b625c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-22 10:56:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2d17a2db5d62e4374e729df75589f7f9c276df0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 13:22:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/363e5c77b89e49e0a0125b6fdb337355 : SUCCESS in 9m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/591094481e71481e9a95328ba79dc633 : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22d2c11f0b0d492ca7ccb412ee43288d : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11dc132d56bf433eb0a6354da643309f : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccbc7dcc591d47a092171b61c14f0ab5 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c15bbfeacb6492093f02e7f1d56ccfa : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a21ca899c1458691e7ee6a1a60fc21 : SUCCESS in 11m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cd7557c356e64ea99a41cf39527d6c95 : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/da78bc93715948eba21498f533cf433c : SUCCESS in 7m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0c5a20cf06f54851874e589627bedd29 : SUCCESS in 5m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/947b3d6f5b7d4fa1b245f37a902fc501 : SUCCESS in 7m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d14b8171c3e84210b1ed9e244e9160a3 : SUCCESS in 1h 04m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d752df5ab5b4130b9fdb77b1bbe15f7 : SUCCESS in 1h 15m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/88ae09ac5e244e66a00652371426365c : SUCCESS in 39m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/290ae27dc5b24829be231b7ac31f562b : SUCCESS in 35m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9eae8114e0fe4a0baa1bbf6296c5c319 : SUCCESS in 31m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3fcce8938b4c4cdda99055409702619c : FAILURE in 38m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/46a8c22f8c994649b5ba249447076aec : SUCCESS in 1h 26m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5aaf1a7004f641589f22881a1a73d3be : FAILURE in 29m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e8ac15a73d0046de93168ad10329dd24 : SUCCESS in 1h 35m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eecde077eb0a4d9eabd098ad3ae9e837 : SUCCESS in 1h 23m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9f4b258296d54ac590ce9ad8ec1ede52 : SUCCESS in 1h 23m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/226e1d050fbb43b3a91994c79e055d24 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee0e5e15772275f6232490635722a14564836a42","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-22 15:05:05.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84dacd4d1eb9558283790e65a7b4520db9d1637a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-25 08:37:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"632fe4340f621893e5a250263e5dd3bf3f9516e9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 12:47:55.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20a6eb1f9e91706995bbfd2227b6c620376b6655","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 12:49:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a1a7fd5b0e907c456e14cf2180170a79f40a5f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 14:55:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0441edf9bf24037b4798d21513e1583 : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14497fcbb6d347e998a19cb82fb1e8d1 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8db9eef4e15d440ba78b1b0b0b894a56 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1b801b33a2f4d8caf2067f452817a87 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f1fd3e4598540918b320c8221d0a178 : SUCCESS in 13m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/54f2139345844e91967f58fbf54f9da6 : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b2e3604a8e904551b7fc70b826ac2db6 : SUCCESS in 7m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fa52456eee0d4f1db7aaddc0313ae6e7 : SUCCESS in 1h 22m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f04851dbea144375a01452c33b126c8a : SUCCESS in 1h 25m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4b80716a684a4e66852f49187958d77b : SUCCESS in 36m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6b3d98f1945248be95ead1ffc346e273 : SUCCESS in 33m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/90f79186eb31409fb7659ff130d9ad1f : SUCCESS in 38m 52s","accounts_in_message":[],"_revision_number":1},{"id":"639183727f5daca4a1c90314221154bb68c3818a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 14:55:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7d3cb0e040786fc7b4e33e0e2bb0920c7537430a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 14:57:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41518f2fc65b482eb2bca7e6d47cdec4 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ee4acb6501c641f19e223b1937c43c0af2fc591","revisions":{"3ee4acb6501c641f19e223b1937c43c0af2fc591":{"kind":"REWORK","_number":1,"created":"2022-04-22 10:56:05.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/44/839044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/839044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/839044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/839044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/839044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/839044/1"}}},"commit":{"parents":[{"commit":"97ac37bc6e6572ce7b368c974b928ce5e35700d1","subject":"Save the HAProxy state outside of its systemd unit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97ac37bc6e6572ce7b368c974b928ce5e35700d1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-21 09:40:57.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-22 10:55:53.000000000","tz":120},"subject":"Fix AttributeError in exception handler","message":"Fix AttributeError in exception handler\n\nThe exception handler of save_state() caused an AttributeError\nbecause it tried to get the output attribute of the Exception,\nwhich does not exist in general. Added a test to cover this\nscenario.\n\nChange-Id: I36acdee1f8782b17c234f6b250facc5c8d0aaf87\n(cherry picked from commit 384b83d88b436c8e657b09f97b12e42d7c9c339b)\n(cherry picked from commit 7aaac224cb4296a4802e4c21c76ba9d010335333)\n(cherry picked from commit 7adeb21c42fdb6431718fcfcd9ad53b8ae077aa5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ee4acb6501c641f19e223b1937c43c0af2fc591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ee4acb6501c641f19e223b1937c43c0af2fc591"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
