)]}'
{"id":"openstack%2Foctavia~839486","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I0f1cac05ebea12a300feffb22fdfe546cf802b87","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-05-03 17:06:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0f1cac05ebea12a300feffb22fdfe546cf802b87","subject":"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners","status":"MERGED","created":"2022-04-27 08:10:49.000000000","updated":"2022-05-03 17:08:42.000000000","submitted":"2022-05-03 17:06:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839486","meta_rev_id":"991ad9eb6bfcc39ef7e811c21be6c522a671f07d","_number":839486,"virtual_id_number":839486,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-03 17:06:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-28 11:44:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-04-29 08:38:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-05-03 14:44:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-27 15:50:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2022-05-03 14:44:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 10:42:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-27 15:50:34.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-04-28 11:44:01.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-04-29 08:38:50.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-05-03 14:44:18.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"1be5cffeb53b6da9d9d5199e4017d4f9b5f07bc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 08:10:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b69f4fd63832bc1c24bfca3a4bc64ae716a2b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 10:42:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43109d1de9b340adbbe24ba19ba01f6c : SUCCESS in 11m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4eea5ff4dd8c43968c0a2727ab6027d7 : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ad0cc42ebec4bb1b2569cb7434a6da7 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b53cc3ce2733468db3ce9af496a119b7 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e491d6d8e7e4c339f166512b1612564 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79f87bd1b1004c47bd8afcfe3fa0d442 : SUCCESS in 12m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7443aa7e7fbd4948b199f79fd6b7faee : POST_FAILURE in 5m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1619b589b9904235a937173001d34968 : SUCCESS in 6m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d803bfe6ed32418d8d92acae8f5a4eda : SUCCESS in 8m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1901f90252804cf39796695b932246ef : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9c3de25d9c8241f68b822d43cefeb87d : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a7ee63d181884feda21a66b9e853e97a : SUCCESS in 2h 25m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd5feb6d4fa946238cedef7d360ef2d0 : SUCCESS in 1h 30m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1a7098117d104708b96e1dcbe47fb1a7 : SUCCESS in 39m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f6e3bb7abbe34a778599fcd2935927de : SUCCESS in 44m 37s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/677cdf11e9b54aa69c697bb06b9e42b1 : SUCCESS in 1h 01m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/381c910700ec44c887ab376b17023112 : SUCCESS in 36m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8b5fbfabf69c45df99123386b9a5c6a9 : FAILURE in 1h 58m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/94e2f97ed53549f69e8f22f7f324ec80 : SUCCESS in 1h 46m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/285ddb6e75f3432c826da39abe033c2a : SUCCESS in 1h 59m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3e58dfe539ba450686666c9cd1fffc98 : SUCCESS in 2h 15m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/579550e301e445dba9c450a5c57830c2 : SUCCESS in 1h 44m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/f49c287a19f84a569283d77f50b2ba49 : SUCCESS in 2h 06m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37c2336f1a9eb44d05ec838b76510039b39cb26c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 12:54:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dcb41877ca9913f97c0b0ed1dcaeb5611b6acb79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:25:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b147e76e55bc4fb681638bf3c675f7f2 : SUCCESS in 13m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ce5f8d153e44585a3ecc1e30cbe2783 : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee04902c2cd440c8bd6f0322e41b39f0 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4de00af8abb4885b00279e23180238a : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e79e79a66afb4acf82b8c63709e149f0 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4f0b776e4fb431187b3118f3f287250 : SUCCESS in 10m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31c7b91802994eb49504136a947d0ed6 : SUCCESS in 5m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0b339e50ffd94d91959ff9342e7ce169 : SUCCESS in 5m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f42f54f913ff4cce8c9e8fc8563216cc : SUCCESS in 9m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f4cb046f34e44059841645d9565784a1 : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9e12e3e5e05b40db9db88eee71d9effd : SUCCESS in 9m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/af7b9e53b1d24c50b73146359b903aab : SUCCESS in 2h 23m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/70a3f12e7a9442fa9a0f99e67f13afc1 : SUCCESS in 2h 01m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1bb157a9b371464b998ef80c96f9151f : SUCCESS in 39m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2dfe09e18f8540fd8f99d8457ceb64dd : SUCCESS in 58m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3a050d0f42814e52a33f90008b040631 : SUCCESS in 58m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be807d01e24d48f0b380157f4bb9cda6 : SUCCESS in 33m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6e794b26a2d546bdbe80f7e9a315e3db : SUCCESS in 1h 25m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/49854fffe1474f53a069c0a4b56a8f64 : FAILURE in 1h 56m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8140c5d1f6f5462090106ee911a8aaf7 : SUCCESS in 1h 48m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4f2e9a38df874d61a620bc3795d291e4 : SUCCESS in 1h 41m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f456397b94024d2391f7c2eec40fd57b : SUCCESS in 1h 34m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/a38eb53b828f48c2890590d973d040fa : SUCCESS in 1h 45m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2874b3f76c53e615ee5b335950500e0accd9fa1a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-27 15:50:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0b799eb37e95ddde149d7bdf0bd34be9f0eb45e7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-28 11:44:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"93093d83358e5c14b3b29d31428bdb48c65162c0","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-04-29 08:38:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"42c60036c71fcf4ef2df3468b1d04d81d7e9da45","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-03 14:44:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a666de0f0c1189fc38cd7b07c344023f8df0e75a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 14:45:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed75356e59bb35c8bed96a25bf8a11bce711d5e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:06:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83bd27b1eae74211bf403f9da995600b : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6f85606d16c4bc3b57e59d52b3a4654 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b8b808af51043c89c372fe85aae189d : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/532f375e895e41fa9647622e9484334e : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22993a3baee94a2a8772d8d335848ff8 : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/397ffeb14af84225b43c4934fe6c3d01 : SUCCESS in 6m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/516967b2ceb14294bdb55ac1aa0c0e34 : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c60b12ebec694db6a8f7446ff08fca03 : SUCCESS in 10m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b27c65ca13c04ea7a16a2c0e2db1c267 : SUCCESS in 2h 05m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a1aa26d1f244fb0b0f589d5f04a6558 : SUCCESS in 1h 51m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/53f808511c62462a836669315b1ab786 : SUCCESS in 38m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f030a32a31d34cb4a19416e2cfce22c6 : SUCCESS in 44m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ea745ffd13fde97cd565829d5dbf486fcbf0bfe5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:06:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"991ad9eb6bfcc39ef7e811c21be6c522a671f07d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:08:42.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d394f446033f480382b399e29661ec4f : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71051a555084413f83a9944fbb0822bf : FAILURE in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4248e98acbfa76ef1ab6d35ffbf0abf5b61e0a33","revisions":{"4248e98acbfa76ef1ab6d35ffbf0abf5b61e0a33":{"kind":"REWORK","_number":1,"created":"2022-04-27 08:10:49.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/86/839486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/839486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/839486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/839486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/839486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/839486/1"}}},"commit":{"parents":[{"commit":"0a062cd66416c6020d9718d87988727bb8a78627","subject":"Save the HAProxy state outside of its systemd unit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a062cd66416c6020d9718d87988727bb8a78627"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-07-09 12:25:39.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-27 08:10:08.000000000","tz":120},"subject":"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners","message":"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners\n\nL7Policies are not compatible with non-TERMINATED_HTTPS/HTTP listeners\n(i.e HTTPS/TCP/UDP listeners).\nThe Amphora driver should deny those invalid combinations.\n\nStory: 2007911\nTask: 40314\n\nChange-Id: I0f1cac05ebea12a300feffb22fdfe546cf802b87\n(cherry picked from commit 45b0a507a150721389833e33789b4cfdbc1a9590)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4248e98acbfa76ef1ab6d35ffbf0abf5b61e0a33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4248e98acbfa76ef1ab6d35ffbf0abf5b61e0a33"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
