)]}'
{"id":"openstack%2Foctavia~839487","triplet_id":"openstack%2Foctavia~stable%2Fxena~I0f1cac05ebea12a300feffb22fdfe546cf802b87","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-05-03 17:06:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0f1cac05ebea12a300feffb22fdfe546cf802b87","subject":"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners","status":"MERGED","created":"2022-04-27 08:13:12.000000000","updated":"2022-05-03 17:08:41.000000000","submitted":"2022-05-03 17:06:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839487","meta_rev_id":"81d08d20ff62c23f58f8bced0b7dbbe03f117337","_number":839487,"virtual_id_number":839487,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-03 17:06:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-28 11:44:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-04-29 08:38:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-05-03 14:43:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-27 11:36:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2022-05-03 14:43:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 10:36:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-27 11:36:26.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-04-28 11:44:34.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-04-29 08:38:20.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-05-03 14:43:47.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"63bfecb1fccef9cb36e01fbd0104b3e580c4952b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 08:13:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4abc7f4e73b51eddff89d81c983137cac5cc624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 10:36:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd769dc1dd2340b0bda10e4f5072377d : SUCCESS in 11m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73b67c86e4484495a911b78eb42865e7 : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1cbf51bb41e4ae583f6d760565773b1 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20095a69c6a84e8998fc8cdebcf7226d : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a64a12da65049608f11f0967507becc : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1084cdeda5284235a7b59afbc246a84a : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dd9b23b8b9b4f97887c9552366a4093 : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1562b4b509e944e9beaf6f2731f363cd : SUCCESS in 5m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/404deda038bc4a9cbbf0a4c4fc4f9b96 : SUCCESS in 6m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/27da84353d274d27b93c6616e487ce1e : SUCCESS in 9m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/87f5403dd8ef4111b05615897ffc45ee : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7215eae14db048448b76d42e42a3f463 : SUCCESS in 7m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/482543190e0d450a94cb7d8833e25b72 : SUCCESS in 2h 15m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d250017e73e4ca7ba2c35d2b44545c1 : SUCCESS in 1h 30m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7e711e62f9374626a45b80570c09c0d3 : SUCCESS in 35m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c5ba7ff7485c4ba18ecce532fb10f532 : SUCCESS in 48m 24s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/c359c740bef54c0daa74d7378d884b90 : SUCCESS in 2h 02m 50s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ec7ed9eccb85441cbf0b1bd68d2446fa : SUCCESS in 48m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c721c848443444c7b0d1e935218623d3 : SUCCESS in 31m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/209dfb0114ce495995d96fba557d42e8 : SUCCESS in 1h 21m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a1ae69446b9348a0a488d471d9b63bdd : SUCCESS in 1h 44m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f6c59064efd74ff382ae07a18a532470 : SUCCESS in 1h 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8eab9507d8e343c5bc34a2ff3d8e77d3 : SUCCESS in 1h 27m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d6edc68aa4344029a577913636e5291b : SUCCESS in 1h 18m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c2b8681915d94aaebaa40a3516855cc2 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d2ede7b9208cf2b5d05d76f65976f4ec90837ec","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-27 11:36:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"02d89b9bf27cfa6b85660879063fff89e812c46b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-28 11:44:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"094ba263b4436e8382e4efc4671569e316e63c0c","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-04-29 08:38:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b21b30bc66c006e6108ca4ed32ee7ef598a8e04c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-03 14:43:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c306f9bbca58983339db888719b0e1ccc99f30ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 14:44:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc7be3858d7c22a8d1d27ffd479ab29eb0250dcb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:06:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd825e3606dd40e4a8dec9ffb62bbc70 : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b2f1105f5ac459192985aa335974e74 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38c3c8c4a70a430baa99043e714b527e : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02afec74bbf74b498e52744637100d30 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bb20caf88fc4bb18b3d4689af2e12e5 : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9820c2b1573744e98771658b7d56efa2 : SUCCESS in 5m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/09599a13523d4ff588c6e2f80ecede29 : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2fa911ece9cc46679bafa19c35fd7544 : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c1cfac6ce09945c4acd1dd76e2975eb4 : SUCCESS in 2h 20m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf7f0c8dc28546c6a27bbf578ec4c52c : SUCCESS in 1h 16m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eb97a898c3b646c4b080c7d023bb3e1a : SUCCESS in 38m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/26bb428fbd324cc18fd7ccb232796406 : SUCCESS in 1h 00m 14s","accounts_in_message":[],"_revision_number":1},{"id":"43405ac37076415a6f1067fa202286fda706e71a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:06:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"81d08d20ff62c23f58f8bced0b7dbbe03f117337","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:08:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52cb1117d15d44aa9c8113718082dde5 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2969fe6373734da69a93ce8dd587b072 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df1ecbda2cc2a32ec34f16d26f23d07453156641","revisions":{"df1ecbda2cc2a32ec34f16d26f23d07453156641":{"kind":"REWORK","_number":1,"created":"2022-04-27 08:13:12.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/839487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/839487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/839487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/839487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/839487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/839487/1"}}},"commit":{"parents":[{"commit":"5b4604c75baba20f1bf3627a4e6291cd77c38876","subject":"Merge \"Save the HAProxy state outside of its systemd unit\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b4604c75baba20f1bf3627a4e6291cd77c38876"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-07-09 12:25:39.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-27 08:13:02.000000000","tz":120},"subject":"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners","message":"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners\n\nL7Policies are not compatible with non-TERMINATED_HTTPS/HTTP listeners\n(i.e HTTPS/TCP/UDP listeners).\nThe Amphora driver should deny those invalid combinations.\n\nStory: 2007911\nTask: 40314\n\nChange-Id: I0f1cac05ebea12a300feffb22fdfe546cf802b87\n(cherry picked from commit 45b0a507a150721389833e33789b4cfdbc1a9590)\n(cherry picked from commit 4248e98acbfa76ef1ab6d35ffbf0abf5b61e0a33)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df1ecbda2cc2a32ec34f16d26f23d07453156641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df1ecbda2cc2a32ec34f16d26f23d07453156641"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
