)]}'
{"id":"openstack%2Foctavia~839488","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I0f1cac05ebea12a300feffb22fdfe546cf802b87","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-05-03 17:06:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0f1cac05ebea12a300feffb22fdfe546cf802b87","subject":"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners","status":"MERGED","created":"2022-04-27 08:13:42.000000000","updated":"2022-05-03 17:08:59.000000000","submitted":"2022-05-03 17:06:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839488","meta_rev_id":"af44daf6973786b3a960507bfad279dad6cda403","_number":839488,"virtual_id_number":839488,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-03 17:06:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-28 11:44:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-04-29 08:38:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-05-03 14:44:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-27 15:49:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2022-05-03 14:44:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 10:10:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-27 15:49:16.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-04-28 11:44:12.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-04-29 08:38:35.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-05-03 14:44:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"bacbeef4a135ad2e819b6641d65f0727793734a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 08:13:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a981f1d35558f7f8c749c6eb8715c47681709b75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 10:10:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e200204c07374cf59b0a18e695db8fac : SUCCESS in 9m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98ae4c616d29410dac405145fe2df70e : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81fee8788eb440c9914348f20fc0ef06 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93b0847e70144a31bcc0f14efab7ef33 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29bc14279b854f779ec955b7a3eb88c8 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/120f1ca3c6ce4eacb18c59c3c9853995 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b2ac25e1484c87b16fa3e7ac19e7d9 : SUCCESS in 12m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77d87865c40b4c3cb79adb4b498b0c0d : SUCCESS in 4m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/164caf86ff904a1eb10dbcaafddefb45 : SUCCESS in 6m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f9b8cd3cd2c643a4bd657c61b67799db : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/072af03e31464bad84e5d1a8dcfe9c85 : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b89536f6767a4b3fb108ba0a6b5cec9a : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2cfc9556c619424782fdf76426a4a93f : SUCCESS in 1h 49m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1c06fc15b3024fa49384f3a9bf464349 : SUCCESS in 1h 19m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d47327c3dcd2472b83fcd1c7c457d2d0 : SUCCESS in 35m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/627b820318174e00b3060432c4c60be9 : SUCCESS in 31m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c1e2a2096634c1cb530dda1cdbb78f3 : SUCCESS in 32m 48s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/33d7df76354b4b118bbb973ee96f5d12 : FAILURE in 42m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7b07452518ba49a084a76d40df87da82 : SUCCESS in 1h 17m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1632d8db77140dab0539407ef7c60cc : FAILURE in 29m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/907752bedafe457486c167ac77f9020b : SUCCESS in 1h 30m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/62c20d343c7446db89208d7db2bf133c : SUCCESS in 1h 38m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/608739d317424726aef1c9990899b933 : SUCCESS in 1h 29m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e6de81f26e684889a7663232e086db70 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"254467ec337a16cfaab06748b1e413f56e3ab419","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-27 15:49:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ed6f129e34e2e639f9938f6ff99e232a45e503b6","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-28 11:44:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"66d9685fdc26dcd823b79be86360f7294fb9c8c2","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-04-29 08:38:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bec9e2cdfac46b7ea8f49309b6fe23575cf0241c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-03 14:44:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"acb46bcca6793c6e5c82993b292d160c75e294ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 14:45:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6d668daa1c41ddfee0deac2f3b8423168ec7a16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:06:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1941c749a514d72960b895a29d8dff9 : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d64d75e8c234ca69cad67e2b8b5d022 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c37379fb46e849139c3b1ab094e9bd73 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20c35aea5c78449894118dce4a92f027 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8a538eb2fee436ea5730b5715f4a89d : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04ab77c9a17a4480b9712a9be5eec678 : SUCCESS in 4m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/691b756156f44671b469cd30dbfe3875 : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c13630d7f0e54e589c09f30291015cb4 : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ac20e3bfd8ef412dafc5d99a12bf717b : SUCCESS in 1h 27m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c8bbb9318e14de2aac0e9a161f6c323 : SUCCESS in 1h 29m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1b231b04adc94d7a8fd508835037e7b5 : SUCCESS in 38m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3cbdb97a3b5c4cc98e9bf20375d5e412 : SUCCESS in 38m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19d56ca0c0d54e14afec4dfa38bf39b6 : SUCCESS in 32m 15s","accounts_in_message":[],"_revision_number":1},{"id":"cd2e585c6e05268ee6e446e21f4c8994b96b039e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:06:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"af44daf6973786b3a960507bfad279dad6cda403","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:08:59.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14231be2465246ebbc6a217e8309af29 : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e761a1e5c3e74edea999081048e60d3a : FAILURE in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d332357d3c4aac693a4be9b0c0b5a4518148d3e","revisions":{"5d332357d3c4aac693a4be9b0c0b5a4518148d3e":{"kind":"REWORK","_number":1,"created":"2022-04-27 08:13:42.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/839488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/839488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/839488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/839488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/839488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/839488/1"}}},"commit":{"parents":[{"commit":"b2f45991fe8627defb7f91d8da41c66e7e1b6733","subject":"Merge \"Save the HAProxy state outside of its systemd unit\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b2f45991fe8627defb7f91d8da41c66e7e1b6733"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-07-09 12:25:39.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-27 08:13:27.000000000","tz":120},"subject":"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners","message":"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners\n\nL7Policies are not compatible with non-TERMINATED_HTTPS/HTTP listeners\n(i.e HTTPS/TCP/UDP listeners).\nThe Amphora driver should deny those invalid combinations.\n\nStory: 2007911\nTask: 40314\n\nChange-Id: I0f1cac05ebea12a300feffb22fdfe546cf802b87\n(cherry picked from commit 45b0a507a150721389833e33789b4cfdbc1a9590)\n(cherry picked from commit 4248e98acbfa76ef1ab6d35ffbf0abf5b61e0a33)\n(cherry picked from commit df1ecbda2cc2a32ec34f16d26f23d07453156641)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d332357d3c4aac693a4be9b0c0b5a4518148d3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d332357d3c4aac693a4be9b0c0b5a4518148d3e"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
