)]}'
{"id":"openstack%2Foctavia~839490","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I0f1cac05ebea12a300feffb22fdfe546cf802b87","project":"openstack/octavia","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-05-03 17:06:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0f1cac05ebea12a300feffb22fdfe546cf802b87","subject":"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners","status":"MERGED","created":"2022-04-27 08:18:09.000000000","updated":"2022-05-03 17:08:48.000000000","submitted":"2022-05-03 17:06:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839490","meta_rev_id":"a168c374c0e4df0e9c403863818eaf6d55c01f16","_number":839490,"virtual_id_number":839490,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-03 17:06:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-28 11:44:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-04-29 08:38:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-05-03 14:45:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-27 11:37:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2022-05-03 14:45:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 10:20:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-27 11:37:23.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-04-28 11:44:29.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-04-29 08:38:28.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-05-03 14:45:00.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"b6b1ef228b18c04a3bab2d2d691c76fe4aea238e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 08:18:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb10c7c11552ae9f00325021bda073afb8e6ef96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 10:20:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2176124d724493599b48d161183bc4f : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ba035ce2e7a492985659ece07d44ac1 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd3ba6ba0fb141679069123d9b2807bd : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae95b0419cda44d58e0cc296f9adf827 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8df15afeec8d4c85877c3bdbc660a1d4 : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9672f5d52ad2409da61228ba0e18c826 : SUCCESS in 5m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1b1a13bde6f64edcb7bb60d07bcce156 : SUCCESS in 7m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/56790dd48b094f2d84ee5de287fe34d3 : SUCCESS in 8m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ab3ade4ec82f44f0af59f2323bf1cad7 : SUCCESS in 6m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/041750dafbad438ea21611e6d490e166 : SUCCESS in 1h 40m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f6b447a629b4590bbc53692ddfdf30f : SUCCESS in 1h 44m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6796109df9f643c08a3513cb7b76e34a : SUCCESS in 40m 55s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2d25ad62a14c45b09283ffa73a10d9ed : SUCCESS in 37m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/89524fbfd48f4066b2a8bd4bdefa7f73 : SUCCESS in 39m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a0dc6a51bb0849f5a34ad9924a7bea37 : SUCCESS in 1h 42m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c4537951317f4afface60e0e0168b66f : RETRY_LIMIT in 7m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b3b70e433c9f4ef391117de5278ae1ed : SUCCESS in 1h 58m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d24655d7549e58919e03956b1df84c1af1bf874","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-27 11:37:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"342cee95f98a8b7f097bb21ca5705690dc3b361d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-28 11:44:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e4da0a24cfb40a6a8b9dfdebcb983bf64a8b3016","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-04-29 08:38:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a39f67cf69bd73bb620eedf15e2f8fe5abb7e856","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-03 14:45:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d389707d36e1761821223fac855bee392e8f2444","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 14:46:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"215cf13a0addd6b45c1ed15811244ae493c770ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:06:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/944e0b441d214ee785241993d68a0134 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7dba068840e145d896d95582e2587c1b : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f4f53f98254433f88e47f66b67bc0ad : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df5e63a8a2574114801bdb385a826930 : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20a966f6b85c4c088fd0c1ab1f2436b3 : SUCCESS in 4m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3511680ffd284a629fe45e7d5a72c1df : SUCCESS in 6m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/56f5896a30524237a873b9955a80830a : SUCCESS in 1h 05m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8beec5d2df504720a249a54d27b40000 : SUCCESS in 1h 23m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/762c3102321148be852de84078d33d1a : SUCCESS in 34m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2745cebd7a6846a082ffad2725822531 : SUCCESS in 32m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb331caa2ee44f83bce179f090ec9cb2 : SUCCESS in 37m 06s","accounts_in_message":[],"_revision_number":1},{"id":"fad11b77daff90c2359d42bfdc7c08dfe452eab7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:06:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a168c374c0e4df0e9c403863818eaf6d55c01f16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:08:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81643688ed00479fa5538101637bf084 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5b9425d94804876865822b9a686e5f0 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22e679a4243cf1fdfd7d50166110299cc8e83fa2","revisions":{"22e679a4243cf1fdfd7d50166110299cc8e83fa2":{"kind":"REWORK","_number":1,"created":"2022-04-27 08:18:09.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/90/839490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/839490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/839490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/839490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/839490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/839490/1"}}},"commit":{"parents":[{"commit":"9838a350a79dac00b8f8184d69ca8cf2acf710e7","subject":"Merge \"Save the HAProxy state outside of its systemd unit\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9838a350a79dac00b8f8184d69ca8cf2acf710e7"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-07-09 12:25:39.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-27 08:15:53.000000000","tz":120},"subject":"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners","message":"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners\n\nL7Policies are not compatible with non-TERMINATED_HTTPS/HTTP listeners\n(i.e HTTPS/TCP/UDP listeners).\nThe Amphora driver should deny those invalid combinations.\n\nStory: 2007911\nTask: 40314\n\nConflicts:\n\toctavia/api/drivers/amphora_driver/v1/driver.py\n\toctavia/api/drivers/amphora_driver/v2/driver.py\n\nChange-Id: I0f1cac05ebea12a300feffb22fdfe546cf802b87\n(cherry picked from commit 45b0a507a150721389833e33789b4cfdbc1a9590)\n(cherry picked from commit 4248e98acbfa76ef1ab6d35ffbf0abf5b61e0a33)\n(cherry picked from commit df1ecbda2cc2a32ec34f16d26f23d07453156641)\n(cherry picked from commit 5d332357d3c4aac693a4be9b0c0b5a4518148d3e)\n(cherry picked from commit 12133d16150917892ebea307701c51f64b638aca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22e679a4243cf1fdfd7d50166110299cc8e83fa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22e679a4243cf1fdfd7d50166110299cc8e83fa2"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
