)]}'
{"id":"openstack%2Foctavia~839848","triplet_id":"openstack%2Foctavia~stable%2Fyoga~Id81ac8dfdd969c7b250d231123d392eff4a86269","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-06-20 13:05:29.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Id81ac8dfdd969c7b250d231123d392eff4a86269","subject":"Fix compile_amphora_details when using UDP listeners","status":"MERGED","created":"2022-04-29 06:07:34.000000000","updated":"2022-06-20 17:38:00.000000000","submitted":"2022-06-20 17:36:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839848","meta_rev_id":"6a367b572dceeed7117125391489a3ad6d36c712","_number":839848,"virtual_id_number":839848,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-20 17:36:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-06-20 13:05:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-06-20 13:05:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-20 13:05:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 08:28:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-04 10:24:41.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-06-06 20:50:20.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"74d995f1f245fe7a27f6ec55d6cf29aa470a1243","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-29 06:07:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5947dddde0225841465a0bcc1275114809d5b78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 08:28:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca16c9ac5d54481fb572f7748d0084d2 : SUCCESS in 11m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f25a526d893a49a9bce17fce0e0c2e99 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61bfab73ba3d48ba906a3f6fb8879377 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d63b76c582a54841ad390909f19d21d1 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/334ce2bb2aaa48859307c3761fe24055 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eccf8834e2f8494aa6d14874faa87086 : SUCCESS in 10m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a006185a49ed46b0a412ff85b247370a : SUCCESS in 5m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a2e735f84c2f4a2fa9d59f623072239d : SUCCESS in 8m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/35da95f0f0a64097b24c88c980851ad1 : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a90c64e24a14c688393a7c745b0f015 : SUCCESS in 8m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d80a14eec164e53b76e3b559cec7971 : SUCCESS in 1h 52m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/523f509c985d4959b2e9a3e5ead8fda6 : SUCCESS in 1h 39m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/66af767b209a43ed9bdac8d76e27e073 : SUCCESS in 38m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f290b7d66fa14d69b13c8d96c22fbc2e : SUCCESS in 56m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9654da1b09514ce28be84d409b0712dd : SUCCESS in 1h 01m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1db6130c7a2749669c20ef1e87701e4f : SUCCESS in 36m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f2499af76e574f7ca7e205f671e6b084 : FAILURE in 2h 00m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4cd57e305142408e9ab61d8037d50236 : SUCCESS in 1h 59m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/195d535f850f4740bd5860dd020f8cd4 : SUCCESS in 2h 17m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/50aa4763236042d09dda14d29da422a5 : SUCCESS in 1h 45m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2272c0647aac40dcb97e767b15cd1bfc : SUCCESS in 1h 35m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/7bbcab969f9044988ad45f38e1e42d84 : SUCCESS in 1h 55m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ac02a676deaf9181f8f74dcfefa745133cdba06","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-05-04 10:24:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3126017fa9c64989904c3b9a06ce3555e6fa87f0","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2022-06-06 20:50:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c036397b1ec372ebc5a3130cda9ee5751045faab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 20:51:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"734c8ec8f712066e0cf6267489c99c346609dda7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 23:30:41.000000000","message":"Patch Set 1: Verified-2\n\n(1 comment)\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fa3232788634231904939d690c18776 : CANCELED in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e296f9b0f2342dab2a6e25c37600e36 : FAILURE in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4bb72eaa7e54ecb99bfc1d0308765bb : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf247de63e09444d944428d396bab32f : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e444a50eec9849ad9f72b3aba3b63cb0 : CANCELED in 7m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/54106b44e6944156b6cc1952bf5fc3bd : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6bdf8fab05a7463daeee2bab7b383eb3 : CANCELED in 8m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/79022bbf0d0b44ea944701774da8c684 : CANCELED in 7m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1744730d20d4d8896eac4ebf950960a : CANCELED in 11m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fea9cfc699474dc2b82783f0dc87b81b : CANCELED in 11m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b1508574e104fcdb2dddb98a72a9c4b : CANCELED in 11m 47s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":1},{"id":"23389a7e59d65fce0c28abec7d2ea7bbd7b71852","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-08 14:48:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"892817a92e64b51bffc5c816979657dce33f4fd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 16:51:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41108e8532714281a890f570736f03bf : SUCCESS in 10m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18903ddc7f3b41f8bf82d272ebcf2a29 : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b5daf4eef9c48f48f8eb43f386084aa : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f95948fd15d4519bb6d084d43b3345a : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6abb959f2f7e485cae389601a056ec7d : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51dd042ad4af4c3a9ad38dd34e64f00e : SUCCESS in 14m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7c2506212b79465c9a9247fbc89b0115 : SUCCESS in 6m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/859ecde1f8894a5a99003682a860715a : SUCCESS in 12m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1a80551ebc384e538783bee4c622fb09 : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a170c58a8fb043999672329792378714 : SUCCESS in 9m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c09037eb548941e3a1027fcd8c8fc7be : SUCCESS in 2h 00m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9cbaf3f6e95e457e9c1d556320c5093a : SUCCESS in 1h 18m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fa71fd1cff0e40b6ae6e573adfa9dfce : SUCCESS in 34m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a8c57475347543a8a38749af09c23eaa : SUCCESS in 47m 08s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b2f2e5ce6c704834b29e1796cf6b0dc0 : FAILURE in 44m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1145522b99f4aa4888b846fb44ff9e5 : SUCCESS in 33m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/779218d76fd3475f9fe388571f01359d : SUCCESS in 1h 22m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/646e85fdd2444fe78f2f8db70f49c3d0 : FAILURE in 23m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/954669dbaadb4669baabc232aa56ba05 : SUCCESS in 1h 38m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ebe241faa19342e8b8f2090d2f54ab93 : SUCCESS in 1h 23m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4cdd650a36e34f27b561e025605f28c4 : SUCCESS in 1h 26m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/4795808067694539bf3e9787c9ae5c5f : FAILURE in 24m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dc65d119262b0c8f41c8fca4d55f1b5071bbcd4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-20 13:05:14.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6ed3aab90d63d1b6a067fc325574ad8c302e346a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-20 13:05:29.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"21797f40463e7b047b5adf22ea0c7b0ae93b5817","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 15:02:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2a477165e21409397e8d7536594d5db : SUCCESS in 9m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b06e3a0818f471c8340df56ae5f1f97 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/895744fe3e744ec3855db4cc97c105aa : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89f4ed1f42a34f6db4c236fd13674488 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e61e81bcd4f1417ea2e199c6f551a286 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9564f8b7f4bf477cab03ec468d8fbb27 : SUCCESS in 14m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/14a351647a404d968a92833186b05288 : SUCCESS in 6m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fe341c2cff984b80b0ae7aa3d506cd71 : SUCCESS in 10m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/eb70e3a47ca7458483b84955e12b69f3 : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9343b6ce0a534551902585fb80035edc : SUCCESS in 9m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e17ee97164c14d03bc081cb21389463d : SUCCESS in 1h 52m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd3cdb24b9024727b5333c3d9eb49dfd : SUCCESS in 1h 23m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/69094f242c1b405a90feca49fa2816b9 : SUCCESS in 34m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/61a246b652a9400181ee9fbdc365fa9d : SUCCESS in 44m 20s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/42d79e286f234d6488527bf2370ae06f : RETRY_LIMIT in 12m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f5ac2f0980f4244a85c4f2206a5a832 : SUCCESS in 36m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dd71b548aa3c4df1b1ef3df1e551211b : SUCCESS in 1h 26m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2e421b8fd33e40819918266d5c8c5465 : SUCCESS in 1h 43m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8c20e567bcc44ec9adac8de619d5f6f3 : SUCCESS in 1h 28m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/09c604d0bb754f84aca23d9dbe434b30 : SUCCESS in 1h 34m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6ba3b69e7a864f488dc6934f172a655f : SUCCESS in 1h 21m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/1496bcb8f53b4326b85d44b6d076fd7b : FAILURE in 35m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"898afac8ba9348cf671e75c07f38d91f8f035536","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 15:02:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"94248afc6d29e10fb81ecc6c2ff4aa1cade9d428","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 17:36:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b48540d5da546d786977d0f1c34f810 : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb711443e9674258a71198a81259924c : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/148a2813d9ad4dc6a2d53e377e712498 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b59fc0218cd7442db8a3d3624a57e7de : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46defb99a69f41e89ace87a0cc8ede5b : SUCCESS in 11m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2c40942f74f5444da0ead89651babb80 : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/679a780bfd244f70923dd9b4a4a7732e : SUCCESS in 8m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/59b02f3313e94d01b7589c1008f67576 : SUCCESS in 2h 28m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d58ef6c7f4074170ad397dbea428770c : SUCCESS in 2h 21m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/446543757aaf479a94720ccd614b3962 : SUCCESS in 43m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cc07d08dbb8341fa82ced751c48f0a8e : SUCCESS in 1h 03m 26s","accounts_in_message":[],"_revision_number":3},{"id":"0e8bc9485a624e5cc66ba5f5fe031c3bac74f61e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 17:36:39.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3},{"id":"6a367b572dceeed7117125391489a3ad6d36c712","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 17:38:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47c3048e9f7041888d3faaa4ec8d4962 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a237182303d6a5559ea1148310a6d8fb18c6cbd3","revisions":{"c44c831649f970d3259bad95d2d2213f9b8167f6":{"kind":"REWORK","_number":1,"created":"2022-04-29 06:07:34.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/48/839848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/839848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/839848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/839848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/839848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/839848/1"}}},"commit":{"parents":[{"commit":"7aaac224cb4296a4802e4c21c76ba9d010335333","subject":"Fix AttributeError in exception handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7aaac224cb4296a4802e4c21c76ba9d010335333"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-21 17:02:15.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-29 06:06:52.000000000","tz":120},"subject":"Fix compile_amphora_details when using UDP listeners","message":"Fix compile_amphora_details when using UDP listeners\n\nWhen requesting the amphora details of a LB that has a UDP listener, it\nfails with an exception:\n\n  AttributeError: \u0027KeepalivedLvs\u0027 object has no attribute\n    \u0027is_listener_running\u0027\n\nThere\u0027s a typo in the _count_lvs_listener_processes function which calls\nan inexisting method.\n\nChange-Id: Id81ac8dfdd969c7b250d231123d392eff4a86269\n(cherry picked from commit 655fdf8c1b5786c3aa24c926dd95537a51a7c0cc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c44c831649f970d3259bad95d2d2213f9b8167f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c44c831649f970d3259bad95d2d2213f9b8167f6"}]},"branch":"refs/heads/stable/yoga"},"2e6043da7bf67fad0f1d821f2f6b6003fc2f21e6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-06-08 14:48:16.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/48/839848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/839848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/839848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/839848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/839848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/839848/2"}}},"commit":{"parents":[{"commit":"01c00f438607576002139afeaf775f641f0a02b3","subject":"Fix new pylint issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01c00f438607576002139afeaf775f641f0a02b3"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-21 17:02:15.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:48:16.000000000","tz":0},"subject":"Fix compile_amphora_details when using UDP listeners","message":"Fix compile_amphora_details when using UDP listeners\n\nWhen requesting the amphora details of a LB that has a UDP listener, it\nfails with an exception:\n\n  AttributeError: \u0027KeepalivedLvs\u0027 object has no attribute\n    \u0027is_listener_running\u0027\n\nThere\u0027s a typo in the _count_lvs_listener_processes function which calls\nan inexisting method.\n\nChange-Id: Id81ac8dfdd969c7b250d231123d392eff4a86269\n(cherry picked from commit 655fdf8c1b5786c3aa24c926dd95537a51a7c0cc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e6043da7bf67fad0f1d821f2f6b6003fc2f21e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e6043da7bf67fad0f1d821f2f6b6003fc2f21e6"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"a237182303d6a5559ea1148310a6d8fb18c6cbd3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-06-20 13:05:14.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/48/839848/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/839848/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/839848/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/839848/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/839848/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/839848/3"}}},"commit":{"parents":[{"commit":"399c2f17504fa156e65c664cc4bfeb1c98f7d9d0","subject":"Fix new pylint issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/399c2f17504fa156e65c664cc4bfeb1c98f7d9d0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-21 17:02:15.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-20 13:05:14.000000000","tz":0},"subject":"Fix compile_amphora_details when using UDP listeners","message":"Fix compile_amphora_details when using UDP listeners\n\nWhen requesting the amphora details of a LB that has a UDP listener, it\nfails with an exception:\n\n  AttributeError: \u0027KeepalivedLvs\u0027 object has no attribute\n    \u0027is_listener_running\u0027\n\nThere\u0027s a typo in the _count_lvs_listener_processes function which calls\nan inexisting method.\n\nChange-Id: Id81ac8dfdd969c7b250d231123d392eff4a86269\n(cherry picked from commit 655fdf8c1b5786c3aa24c926dd95537a51a7c0cc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a237182303d6a5559ea1148310a6d8fb18c6cbd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a237182303d6a5559ea1148310a6d8fb18c6cbd3"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
