)]}'
{"id":"openstack%2Foctavia~839852","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Id81ac8dfdd969c7b250d231123d392eff4a86269","project":"openstack/octavia","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-06-20 13:06:06.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Id81ac8dfdd969c7b250d231123d392eff4a86269","subject":"Fix compile_amphora_details when using UDP listeners","status":"MERGED","created":"2022-04-29 06:17:36.000000000","updated":"2022-06-29 16:22:14.000000000","submitted":"2022-06-29 16:20:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839852","meta_rev_id":"31e4932055efe106e198ad1536d2cdfd22fccb13","_number":839852,"virtual_id_number":839852,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-29 16:20:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-05-04 10:24:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-06-06 20:49:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-06-06 20:49:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 08:02:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-04 10:24:54.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-06-06 20:49:44.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"36bac5af72e792d5a38e72653d1b3c775c261d65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-29 06:17:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46d0028c67bd275b8979acfa422083e68e687fe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 08:02:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7c70ba9e2384678a598b268eb95c158 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f3c3e68e3e149e8be859b2e3637463a : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2946a700bd3417e80a6dc1fa5374862 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/decefe0b61a14810a974f827e98d83bd : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66863e98e94543858ed8187aa259d83f : SUCCESS in 9m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/32ef23979de2404e9334fbaaa15795de : SUCCESS in 5m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e0aba4163d74484fb2eb0113d7b04bc8 : SUCCESS in 8m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9c7ede49c9bc4d01a3a169ab172c7500 : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4e274b2032bc4bb2a0529c74d9a24b77 : SUCCESS in 1h 07m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4bc08c1e4fa145cb83cfc83e7a081675 : SUCCESS in 1h 42m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2c6d917dfb1549e88d00ebcf730a5314 : SUCCESS in 36m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c6138c9d30b3453c8b163ee4b9d9d28d : SUCCESS in 34m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/52e4c6e337dd4536bf8aefd614728d0d : SUCCESS in 35m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9b1829a1acfa437e87f02c7a547d5697 : SUCCESS in 1h 31m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b1ac488b70b54655a6360c196157da57 : RETRY_LIMIT in 6m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7931cbe953fd4604ad7f6c39188f0f66 : SUCCESS in 1h 39m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fca4100c702434a3dfd267b16e6b062f201aabf5","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-05-04 10:24:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c204de1e8c47a18b09a08ca9cfffeaa67c630d2f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2022-06-06 20:49:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"311247222a5fb99c465f2f8eb2cba6e6047edcb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 20:50:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d842a6b98b142f1c2e039578431fae8b10f5ca91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 21:20:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db15daba153f49c6ae6dad61ec8b84da : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae7cd5330fc54994b0aef4cb652a48f6 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86535eb0343b4ea98e16300993263c4a : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5441104392c849a9b469bcc2829022b0 : SUCCESS in 11m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/29799f216a7b4344b63257f9f063f7cd : SUCCESS in 6m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8e172acba9814677a3b00275b81d5346 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa695189aa2143fc81474b30a4afb62b : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8c5add64e97c4d79ba73d5248dbe8dfc : FAILURE in 28m 24s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/427e5d83184544ba80ed38a9c6b69599 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ffece2d482b4e68a066cb9f857609a8 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"f2b4080f3cbc9ff1169fae8e8f81c6fc764db1df","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-15 11:20:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c6aab70896f801294b646b920fb4163671f4f61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 13:02:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd64701699e942cf8b43ae4ffa880f1d : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09f69f14c4524ca58febee6a8ce68d2f : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a81031ff10264b16aaa0282a0a0d2953 : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72a3d5ec099349db8cef95a78f9cc957 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67f9bd1ee8dc4fa78adfd779a2feb273 : SUCCESS in 11m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6147ff36efda424f955b56b127ffb9ee : SUCCESS in 6m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cfcc419639074ba5b9ce1f54b1f75a28 : SUCCESS in 8m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c94d1e89e384f809e24f455df130dc2 : SUCCESS in 6m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/16f9705c926b4f539bb929a27422aa9d : SUCCESS in 1h 06m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9837bbc146c463b978a648485213b65 : FAILURE in 30m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/df2a273b9c5e4df294a8e060119ebe77 : FAILURE in 28m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a7b8a22a1f0349b986c459ed3617d035 : FAILURE in 41m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96f4b8c3aa7f4ec2b14f3338e33ba441 : FAILURE in 26m 21s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/49425aab9d9049d0a2bcdf851c146e10 : FAILURE in 30m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e2ef9632a56540799159a0363c103386 : RETRY_LIMIT in 7m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fb9a23cf33194b2187c477cffb89240e : FAILURE in 29m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16692604a72b3d9b236ecf42b11c8289b0559fee","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-20 13:06:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2acf6570fb727c7a1d191c50e504258bf806ace9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-29 12:54:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a639008e3f131ac00262ccaa97d9aaf63653fd3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 14:34:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2346988e6ecb454f8e31ffdb31be98f3 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46319a891ac24b808d8ae818011291fa : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7df78251e084e19a09b59b8c5237f8b : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/781a02edba1a432e86e47286d2dc61da : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf85612292ba46a98c3b9086aafa81c9 : SUCCESS in 12m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bcd46f414a494647b67a2158cd013db7 : SUCCESS in 6m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a526c48b06454c30b0b625b958d32dde : SUCCESS in 8m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c2ea54e9a4de4914801145ee2bfc01ad : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9013f81cae5447c2b215c87180ec4aeb : SUCCESS in 1h 07m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bda232e33d434638993e9f6a57ff0f1c : SUCCESS in 1h 29m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/83277afc721840bc9b85bd7a8d114705 : SUCCESS in 48m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c9316ab98915454da7bb031374d6d8a5 : SUCCESS in 35m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e13284ceb5754ef79d13a5f116f485e5 : SUCCESS in 36m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4492be538e714c21b11adbd9703b7a7a : SUCCESS in 1h 33m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a26815c3bf7344a79c942d33324bd1ce : RETRY_LIMIT in 6m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5112f845abda4be9b9aba2f0a7c64163 : SUCCESS in 1h 37m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc9beaef43bb66d906e2e24a81c25ad04b3f3261","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 14:36:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c19340d750366b1e6a0643ba2a2995dfc6204a04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 16:20:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e08227c0866d40418a8ff18bcda7619a : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84c28c8ef82141c99d117f2ccbec853e : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/74f9bc042cc1434f9de96f0afaf5220b : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e522073ba79549db85c73502dcc549a0 : SUCCESS in 12m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0ad25e631f3246b2b86d35c8da039aa3 : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2a128608067a4f229a7bd6cf5c6fc4a6 : SUCCESS in 1h 04m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/913936b5aa104cd6bc3b8842e82663bc : SUCCESS in 1h 42m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/40f5aadafa1246bd92a2818e14b59cce : SUCCESS in 35m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/94679234b2074f54ba67aecbbe9f3dfe : SUCCESS in 34m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b5a6fede8074cc6bc4e20a841633763 : SUCCESS in 35m 12s","accounts_in_message":[],"_revision_number":1},{"id":"2662d769cf6dfaf972124291f2ae5cbfefc00f90","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 16:20:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"31e4932055efe106e198ad1536d2cdfd22fccb13","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 16:22:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5212cbc8f7641ea91c87819722099b3 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9cb77a0b81e7dda1cb7e9240fc00c122dcdfc48b","revisions":{"9cb77a0b81e7dda1cb7e9240fc00c122dcdfc48b":{"kind":"REWORK","_number":1,"created":"2022-04-29 06:17:36.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/52/839852/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/52/839852/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/839852/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/839852/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/839852/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/52/839852/1"}}},"commit":{"parents":[{"commit":"be6062bfa4a5fe0194321f58375257b8fe1a14b8","subject":"Merge \"Add keepalive for redis-based taskflow boards\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be6062bfa4a5fe0194321f58375257b8fe1a14b8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-21 17:02:15.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-29 06:17:23.000000000","tz":120},"subject":"Fix compile_amphora_details when using UDP listeners","message":"Fix compile_amphora_details when using UDP listeners\n\nWhen requesting the amphora details of a LB that has a UDP listener, it\nfails with an exception:\n\n  AttributeError: \u0027KeepalivedLvs\u0027 object has no attribute\n    \u0027is_listener_running\u0027\n\nThere\u0027s a typo in the _count_lvs_listener_processes function which calls\nan inexisting method.\n\nConflicts:\n\toctavia/amphorae/backends/agent/api_server/amphora_info.py\n\toctavia/tests/unit/amphorae/backends/agent/api_server/test_amphora_info.py\n\nChange-Id: Id81ac8dfdd969c7b250d231123d392eff4a86269\n(cherry picked from commit 655fdf8c1b5786c3aa24c926dd95537a51a7c0cc)\n(cherry picked from commit c44c831649f970d3259bad95d2d2213f9b8167f6)\n(cherry picked from commit 5997304d878d63adf495b470993674f0e0fbcc9a)\n(cherry picked from commit 2b2494d7c737843c759ca5aaeb8edf17eab05d01)\n(cherry picked from commit c362bacfe1d123467d4a3a1b987c03bf14dbb37c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9cb77a0b81e7dda1cb7e9240fc00c122dcdfc48b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9cb77a0b81e7dda1cb7e9240fc00c122dcdfc48b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
