)]}'
{"id":"openstack%2Foctavia~839853","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Id81ac8dfdd969c7b250d231123d392eff4a86269","project":"openstack/octavia","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-06-29 12:54:22.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Id81ac8dfdd969c7b250d231123d392eff4a86269","subject":"Fix compile_amphora_details when using UDP listeners","status":"MERGED","created":"2022-04-29 06:18:08.000000000","updated":"2022-06-29 17:23:20.000000000","submitted":"2022-06-29 17:21:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839853","meta_rev_id":"6a523ac1aa31e4f41ee884de2ed7a02e148c3d96","_number":839853,"virtual_id_number":839853,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-29 17:21:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-05-04 10:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-06-06 20:52:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-06-06 20:52:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 08:01:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-04 10:22:24.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-06-06 20:52:38.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"bcf249ed995ddfe98309514b64a3485562cd08c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-29 06:18:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fc8185cd602c22aba07584c373143823abf3517","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 08:01:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/904bd834ce3a478d8b74df8c4bf16b47 : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c0bf092cd9b436c8c01604b9689f4c7 : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/565521b243af42148ea5872635db8ad5 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dd6682c766544ff9c11ebd95e2c60b5 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/403bf3b9fb8945ac8774387902a10ffe : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/669bdfa5bb344a2e9f51c4c137b25e9c : SUCCESS in 10m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ed2ae07a7f674e50b6b01ffb1f52c842 : SUCCESS in 5m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/04bf6f9c5bd044ac97b9b5cce3daf28e : SUCCESS in 8m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4525739fa3ac4367bc00d767cdb97bdf : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d8237907b7db4254bace3749b2747448 : SUCCESS in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/22388f5cded244b1807a1c8e94d4a2c9 : SUCCESS in 1h 05m 34s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/59bf33181c7040979405f2d43d840809 : SUCCESS in 1h 10m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b09c39f7f2d64b89ab96479e7b3f387a : SUCCESS in 1h 40m 33s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/87ab6b69163a4b5a9919b03867465e9c : SUCCESS in 1h 34m 43s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c3b6ba6fa10e4dd1ae99f4dd70abeed3 : FAILURE in 16m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0fbcb962b0474707b0cbd3c43e563c2a : RETRY_LIMIT in 6m 33s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/b734e92f441f4b99841a2d5d75ea24ad : RETRY_LIMIT in 6m 57s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/381657df3bb245a3b0d22ff28462e704 : SUCCESS in 37m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b70f8ef7d6014cccbb1948cfcac8baab : SUCCESS in 1h 34m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/548469539559424bbc7ad202b9b53435 : FAILURE in 35m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fa11ea4116894dfdb4f39fadc56a394c : FAILURE in 33m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d5807d712d7e4658a8df9e02de156091 : FAILURE in 14m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1f08c6240ce8876b02ae4f85883104c8fd3f153","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-05-04 10:22:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1eb77867d1e1baa06cae40a9273b0ee184fcae80","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2022-06-06 20:52:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f66a59e9d62640a421b8e6983106bb1e02d721b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 20:53:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7952095fa2af8c6b4b391c74f5ab8af7ba131f16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 23:40:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69adf74ded8f463194fa76269423ba37 : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e48980b0033d48e4b25382905f93ebfa : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fec5d6fa23284080afece2f186f38be2 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/859841b3492d4fdba1a776334af95445 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81d745a92c4f417db6c642e7ad55d982 : SUCCESS in 12m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8eb43e46de2d4288b560f5eae1087e1c : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5300eb2f4ef54736a901ebaceb61810f : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cf4199a2153343d6a956f863f009cbaf : CANCELED in 18m 23s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c3f5668818974a61bfe7a261ec2b181e : CANCELED in 18m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/35476203dd374607bdced9f6962ac74c : FAILURE in 18m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3373df8d55f5dcd658d8307b23d2a884b0b33f14","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-29 12:54:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bff3c7fb44042394667321b7a0fff731659d14b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 15:12:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/182952c2827b4a89b0deb3584fb6c4c8 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b62d521d3570402f8d71af129c982050 : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b380bf30689e4c9db1431af4d86e23ea : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4f452c2da3c46c08457a5767adcf85d : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/153bffed18a44049b15cb55ab019afa9 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ad2ce4e20724d9193c77b6239f7a727 : SUCCESS in 12m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e79cce5c4fe146f5b9e0be5782022bdc : SUCCESS in 5m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/21fa0c283f5441de9e283ea60484496f : SUCCESS in 7m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/65f468829c62424299cca09e757d3c5a : SUCCESS in 7m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c1664bf5dfc04e6da5c6b00290b97e95 : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4d34b14e85f840b195c562f653ab0d8c : SUCCESS in 1h 35m 54s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/71954f373ca34f7a8e84aef7185a7171 : SUCCESS in 1h 37m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d7f2e58e8254e188054b5c24a6933f8 : SUCCESS in 1h 57m 03s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/027242f8b3914fd1a9e13643e939d179 : SUCCESS in 2h 05m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/49a4ff5168014657b7a372d2d56206ac : FAILURE in 12m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8980d83d174b40e6b6ff529e55094e53 : RETRY_LIMIT in 6m 49s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/8f27d6b9658349f39098fe31388b1fa1 : RETRY_LIMIT in 6m 59s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3b9df93a6ac840429521c5b6836ff8bb : SUCCESS in 40m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/91f403dc2def416f951da74461833fd8 : SUCCESS in 1h 42m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f229ef1ed494aba9961e50bf6ebfbde : FAILURE in 36m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ddc6c5b107a04df69a85bd41dcfc008c : FAILURE in 34m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1235aadf8a9c47bfb5d35d3874a31bb8 : FAILURE in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"606f112a8d2d914b543dbea66bf12923772326ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 15:13:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"90d8c90ad64cd52b9206e6b0468bb7d7da39cb32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 17:21:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/313664511de3482183a42aff8b702fd8 : SUCCESS in 6m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7aff86e25c4426689260dee7b063cbe : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b87b9b92c2641e49dc8d30c5268d6cb : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/299aa3bc971c4b5c886d1e009842bc65 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20cecf394f9e4ff9bf46d047576d4fef : SUCCESS in 12m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/24cc24d2f4d84ca5997f34160f6f577d : SUCCESS in 6m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c4ca508f757c49348a52f51c06f8ce7a : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/56d67a027bd346098e7722babda0e819 : SUCCESS in 1h 36m 18s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/986b25bf3f6d4902b17b3fec06901b21 : SUCCESS in 1h 37m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f05d4d74c7c4d4c800a53ed0bacea04 : SUCCESS in 2h 07m 04s","accounts_in_message":[],"_revision_number":1},{"id":"4ff89dfcee7ae85e50bf3ae99f4f3ce88ac1312b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 17:21:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6a523ac1aa31e4f41ee884de2ed7a02e148c3d96","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 17:23:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/772713b0406f43829a0ffcf2f2a8fca5 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3637fe83fbd1ae643410e072aa1a7fd1b070a1b5","revisions":{"3637fe83fbd1ae643410e072aa1a7fd1b070a1b5":{"kind":"REWORK","_number":1,"created":"2022-04-29 06:18:08.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/53/839853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/839853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/839853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/839853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/839853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/839853/1"}}},"commit":{"parents":[{"commit":"d95152b7d3c8fff286fd29fa339578dcf7226ddb","subject":"Merge \"Save the HAProxy state outside of its systemd unit\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d95152b7d3c8fff286fd29fa339578dcf7226ddb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-21 17:02:15.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-29 06:17:45.000000000","tz":120},"subject":"Fix compile_amphora_details when using UDP listeners","message":"Fix compile_amphora_details when using UDP listeners\n\nWhen requesting the amphora details of a LB that has a UDP listener, it\nfails with an exception:\n\n  AttributeError: \u0027KeepalivedLvs\u0027 object has no attribute\n    \u0027is_listener_running\u0027\n\nThere\u0027s a typo in the _count_lvs_listener_processes function which calls\nan inexisting method.\n\nConflicts:\n\toctavia/amphorae/backends/agent/api_server/amphora_info.py\n\toctavia/tests/unit/amphorae/backends/agent/api_server/test_amphora_info.py\n\nChange-Id: Id81ac8dfdd969c7b250d231123d392eff4a86269\n(cherry picked from commit 655fdf8c1b5786c3aa24c926dd95537a51a7c0cc)\n(cherry picked from commit c44c831649f970d3259bad95d2d2213f9b8167f6)\n(cherry picked from commit 5997304d878d63adf495b470993674f0e0fbcc9a)\n(cherry picked from commit 2b2494d7c737843c759ca5aaeb8edf17eab05d01)\n(cherry picked from commit c362bacfe1d123467d4a3a1b987c03bf14dbb37c)\n(cherry picked from commit 9cb77a0b81e7dda1cb7e9240fc00c122dcdfc48b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3637fe83fbd1ae643410e072aa1a7fd1b070a1b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3637fe83fbd1ae643410e072aa1a7fd1b070a1b5"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
