)]}'
{"id":"openstack%2Foctavia~839856","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I433e074c2eac5f6875ee5dace8faf86969b24f66","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-06-29 20:26:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I433e074c2eac5f6875ee5dace8faf86969b24f66","subject":"Fix potential race conditions on update requests in the v2 worker","status":"MERGED","created":"2022-04-29 06:45:14.000000000","updated":"2022-06-29 20:28:17.000000000","submitted":"2022-06-29 20:26:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839856","meta_rev_id":"e349d2ae9a3651e2ac3e28ab0b95f2f9b774a4e7","_number":839856,"virtual_id_number":839856,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-29 20:26:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-06-27 06:53:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-06-29 17:02:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-06-29 17:02:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 08:29:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-27 06:53:19.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-06-29 17:02:34.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"af5771aacab3113511ebaa2f833fd47893577798","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-29 06:45:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3d4d4cc1899803f3c93f09201239a789aa9f2ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 08:29:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ddbbca5a91d43e89c5e0b1f1f5f1e47 : FAILURE in 8m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1606b61b150b4d879c21542975b4bbcc : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03b456f64d0b4f078108777322610671 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bedaf68160c44a49892424355ad533c : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d556a8f723a44dda0dc080b6f6dbcfa : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a755ffb8391043c5b05e554d06d9071b : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ebbd17746374008b496a05597ddfeaa : SUCCESS in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9061edbec0444af991f315590058598d : SUCCESS in 4m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/858ef62e60014e7aa19f757f6581a93f : SUCCESS in 5m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c4f970946c084d958cbf358da8e3163e : SUCCESS in 8m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/69e4aee53a7c4a1297bb3bacfa010460 : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c976f27f7b934f869425273ae43fe666 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/41c731cad058442297e95ee2d11f5919 : SUCCESS in 1h 42m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c9baa034507a43f6991f572c9712721b : SUCCESS in 1h 29m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a43e3685fd3447d18b72795624492206 : SUCCESS in 1h 13m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2481cf78b0f04b1f800fbbad6741acf7 : SUCCESS in 45m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c971328acf2641ff9e023ef7373c3981 : SUCCESS in 40m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c3d8f52cfa0e46c3bc3fdb58cac1123b : FAILURE in 1h 13m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/67087a91369049c78401a00962ef99ef : SUCCESS in 1h 33m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/56ad4616eeb141bd941146a5998d2530 : FAILURE in 31m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/12dfebada6cf4d9c9de1e7afec1af27f : SUCCESS in 1h 33m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3f7dfcd7de9c47fb9cbe52d4b726988e : SUCCESS in 1h 40m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/68e1067547a34d619cda5b49677133eb : SUCCESS in 1h 18m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/68f0a8de42d54daa93eea7430f5f08c8 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6b170bf39a5cb54894243981985d64f290971d3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-02 09:23:45.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"12fd430317bb28099c66acf66dc961331112dc11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-24 09:58:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"85e41c37417862638d6df28159c32373afc493f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 12:02:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35e6c6dbf3f6493180133745fd7cc74e : SUCCESS in 11m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51c57b48cb9940399c96095f956b3942 : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/766142bdb9784026a3b538e52f004917 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc0e717a36e5452788a52fd3b077add5 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/153d0fba3f7640c7b8c25834d109b6b3 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d2b2649a3ed48ddaf350ceb455e9141 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba16463f885c47e3a4e17ee650d33015 : SUCCESS in 18m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cabfd17329644d508052687778a9c297 : SUCCESS in 5m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7b91471b6abe4dcdb8a53e8fac64e261 : SUCCESS in 6m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/91fff4857b60422aa3f25fe8bd4d9c22 : SUCCESS in 9m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/68accafef8f6499a81f68557c7d9cc6e : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a4561cce335c4f64a088bca52590ac92 : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2cc41c2d73be4a079d750a9493784975 : SUCCESS in 1h 37m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b0fa9b1c8614223a685e1aece29efa9 : SUCCESS in 2h 00m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/acac4ec09d9348e6939caa6de1d27eab : SUCCESS in 35m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3d81d6bbca3b4975953493c1ba282c94 : SUCCESS in 34m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6533646b6f54bbfaf479dc70d282e0a : SUCCESS in 42m 52s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4ceedb212711438bb7a71c9336b3ca28 : RETRY_LIMIT in 13m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1caa69efb118498ba8e35c9d32715e05 : SUCCESS in 1h 49m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6fb219615c3f424ca88896510afa9551 : FAILURE in 29m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6f3440a97c9b4130a9d7b960b1ba6507 : FAILURE in 1h 55m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d813205b42a24716a0af45bece8d03d7 : SUCCESS in 1h 42m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d0a77419b38b41a99786aa6a1c487da6 : SUCCESS in 1h 34m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/61e21470cff24fef9b8fd2e3c554a290 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6d011fb658a4ebaca7316d6f2366d12caf631a79","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-06-27 06:53:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4a43a030e526815412e7f1785a884aaefd331c59","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-29 17:02:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bbdc14489eed8ec42f60d83126de8c29019d7c31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 17:03:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b3022a7c94e4bea2b92a2428f5e54bfc0f028229","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 20:26:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/978598400e11448ab79f9800d95c8567 : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee410ff3893b406985f54808d8a5b5b8 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/233533a3006946ddb56b393f5db89fcd : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a81af3948c34d9fba8678dd2c3a57b8 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/673c513394a5477f83877c78eb1ca5ce : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/622a2e3a745e4ac4a02e66af551b2af7 : SUCCESS in 5m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5c1f30a50d2646a6a1462a161d7551c7 : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/400e92f237664ca398f03e4521f0cfaa : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/55c52954b1194c3fa997e2cfe0eb401d : SUCCESS in 1h 26m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21e37406085547b0b66177d1a1497b76 : SUCCESS in 1h 19m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d94aa36e7e13413c867100c0999cbae7 : SUCCESS in 43m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d2518f499cf84675990e99b174295a70 : SUCCESS in 31m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d02c2d0a43241eda547fb711d596b75 : SUCCESS in 33m 41s","accounts_in_message":[],"_revision_number":2},{"id":"442fd082f0b7b768759b7c1e16bbfda2a11aff69","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 20:26:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e349d2ae9a3651e2ac3e28ab0b95f2f9b774a4e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 20:28:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e426fde01c8e42bfb6c6471655d06627 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfbe595973f345d09344b6d8095a0f88 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8a7ece8a904d09afcc11c696570495b1594a0c08","revisions":{"331b11c10532b097829808c3a18382abff56ed45":{"kind":"REWORK","_number":1,"created":"2022-04-29 06:45:14.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/839856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/839856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/839856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/839856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/839856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/839856/1"}}},"commit":{"parents":[{"commit":"69486c81bd5ad5adcc8534c682f834a6c5c400f4","subject":"Merge \"Add keepalive for redis-based taskflow boards\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69486c81bd5ad5adcc8534c682f834a6c5c400f4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-08 19:25:48.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-29 06:44:48.000000000","tz":120},"subject":"Fix potential race conditions on update requests in the v2 worker","message":"Fix potential race conditions on update requests in the v2 worker\n\nThe v2 worker was not waiting for a resource to be set to PENDING_UPDATE\nbefore updating it effectively. In some circumstances (loaded\nserver/slow DB for instance), the resources may have been updated with\nold data from the DB, resulting in a no-op update.\n\nThis wait for a PENDING_UPDATE state exists in amphorav1 but it was not\nmoved to v2 when we switched the logic from the DB-based objects to\ndict-based objects.\n\nStory 2009887\nTask 44645\nStory 2009985\nTask 45032\n\nChange-Id: I433e074c2eac5f6875ee5dace8faf86969b24f66\n(cherry picked from commit 6042de759db557f421482d22185c5e5075511417)\n(cherry picked from commit c31051b3881b80ec706161c92a3929e8cc9d3d18)\n(cherry picked from commit 0e4efed23cb688867025a866565e371f8fc477ca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/331b11c10532b097829808c3a18382abff56ed45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/331b11c10532b097829808c3a18382abff56ed45"}]},"branch":"refs/heads/stable/wallaby"},"8a7ece8a904d09afcc11c696570495b1594a0c08":{"kind":"REWORK","_number":2,"created":"2022-06-24 09:58:33.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/839856/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/839856/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/839856/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/839856/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/839856/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/839856/2"}}},"commit":{"parents":[{"commit":"69486c81bd5ad5adcc8534c682f834a6c5c400f4","subject":"Merge \"Add keepalive for redis-based taskflow boards\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69486c81bd5ad5adcc8534c682f834a6c5c400f4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-08 19:25:48.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-24 09:52:00.000000000","tz":120},"subject":"Fix potential race conditions on update requests in the v2 worker","message":"Fix potential race conditions on update requests in the v2 worker\n\nThe v2 worker was not waiting for a resource to be set to PENDING_UPDATE\nbefore updating it effectively. In some circumstances (loaded\nserver/slow DB for instance), the resources may have been updated with\nold data from the DB, resulting in a no-op update.\n\nThis wait for a PENDING_UPDATE state exists in amphorav1 but it was not\nmoved to v2 when we switched the logic from the DB-based objects to\ndict-based objects.\n\nNote for stable/wallaby: this backport also contains [0] which improves\ncode coverage.\n\n[0] I403e1861b1b51c07edad35c64ffac22167342c88\n\nStory 2009887\nTask 44645\nStory 2009985\nTask 45032\n\nChange-Id: I433e074c2eac5f6875ee5dace8faf86969b24f66\n(cherry picked from commit 6042de759db557f421482d22185c5e5075511417)\n(cherry picked from commit c31051b3881b80ec706161c92a3929e8cc9d3d18)\n(cherry picked from commit 0e4efed23cb688867025a866565e371f8fc477ca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a7ece8a904d09afcc11c696570495b1594a0c08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a7ece8a904d09afcc11c696570495b1594a0c08"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
