)]}'
{"id":"openstack%2Foctavia~839937","triplet_id":"openstack%2Foctavia~master~I2f49cbb7f8d99ba2da878bbfc7081a3cc3b3aa07","project":"openstack/octavia","branch":"master","topic":"amp-timezone","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-05-06 16:52:03.000000000","reason":"Michael Johnson replied on the change"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-05-17 11:58:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2f49cbb7f8d99ba2da878bbfc7081a3cc3b3aa07","subject":"Make amphora timezone configurable","status":"MERGED","created":"2022-04-29 12:20:05.000000000","updated":"2022-05-17 11:59:56.000000000","submitted":"2022-05-17 11:58:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839937-amp-timezone","meta_rev_id":"bea7e035120cb5f4ee86274c0206e9bd3d9efe13","_number":839937,"virtual_id_number":839937,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-17 11:58:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-17 09:40:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-05-06 16:52:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-17 09:40:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"disliked":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-05-17 09:40:49.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":-1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 15:37:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-02 10:02:37.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2022-05-05 14:03:03.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-05-05 23:00:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"bbc345f705e5a1ec0a67c7eb7063e02fea723e20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-29 12:20:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f68572fbc819bdc905c339b72a9e35e43c60747","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-29 12:55:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"22a34ef968ebab42bb64eb43605743102bcc06a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 15:37:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7909dfb891d24018b0816c39e37d78d5 : SUCCESS in 14m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/426d1c650b02440c90192eff348f1abd : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/188719c394c340dea68173d228e43998 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba84a331cfdb4a7293f0ec155cbfb46b : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40fe58f094dd4a87955bfadbce55cb70 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4398ea4e539d424da1352eaaab0d9206 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82a06a9ed53d47fdad9b399b48df50b1 : SUCCESS in 13m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/144937530a2a4cb58ef793a4cb1157bc : SUCCESS in 5m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/03b7ab2a80d541359eb1f93af0d65364 : SUCCESS in 6m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e056392bc518493497153b009bdaade3 : SUCCESS in 8m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/91880825ab20418f969498c5cb6311aa : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/be2a4dd9dffc4795b42db8f321da1548 : SUCCESS in 7m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b5429c7a485b419aa1fe5a0c7e715f12 : SUCCESS in 2h 34m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db1e1590b3794caba9b148838fbffbb3 : SUCCESS in 1h 33m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6afdb8217744413c89d57fa41c503974 : SUCCESS in 34m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1adb4d1038e347d2beb6bac0f5cf1baa : SUCCESS in 52m 43s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3a783c3741ca4cfaa1eae330f022f659 : FAILURE in 28m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4542830d73a94b3f851505e9c5c4876b : SUCCESS in 34m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/17f885b6b9334c2bae294750ae554c59 : SUCCESS in 1h 37m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/62c43fccc2be4cdfb6ec63d4a9c36a4b : SUCCESS in 1h 48m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/875175b09e40448c850697df79e86615 : SUCCESS in 1h 37m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5b85737f45ff4552a3f8f23df6a55bce : SUCCESS in 1h 29m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2a07e3ab3c30431aa8d5ea1157c6049e : SUCCESS in 1h 25m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b784e87f55e4a6e9c2077eb66ea3428 : FAILURE in 16m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f71f98970cccc63fb6425d00547156e4bf3833cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-02 09:47:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"692369cd4c35652736fc0e7085d8365b5294bc2b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-02 10:02:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"269c88e0574cffe4b0eaf8472259b4399cc880b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-02 11:12:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"63c54a80f61d8732ac991255e437c183e374d37f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 13:20:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f282a4aa662480098072ee71d29299a : SUCCESS in 10m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7e624485ec84022a17cd8dcb31aaa2c : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21b9faac3a7a4a47ac7af9ecb3b0619c : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f912393c7d724ec29832c6b147eb69f4 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6554c0fade44bdda6b4af7c6d26a790 : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00818e9e4be54f9eab35809c09bcc4c5 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac061e483fed48178cd31078c6105cbb : SUCCESS in 13m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f254ecee8d34649b8564fad90797ced : SUCCESS in 6m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/10a67e3eaf6443b59de3987f0e91a03a : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/87f8a77e9d84435791255bdaa29119ba : SUCCESS in 8m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/635dd405f39e47a8b69e852534b65aae : SUCCESS in 4m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9dcd3c1ba242475a9abb91c999ef3bef : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/63e22951b0924f47851e644043b1ea48 : SUCCESS in 2h 04m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49f7741434fb4a9bb7d8ccf56afd6b16 : SUCCESS in 1h 57m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/37298bc68f4b4d0aa34e173b9190bdab : SUCCESS in 39m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bd28290320ee4c0f87e103d50a436337 : SUCCESS in 51m 25s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e4b6f441b6c943fcb0c615440b3c5162 : FAILURE in 19m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2cd4e22dd3a0484cb379cc210f6cd070 : SUCCESS in 36m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fbe66f9b20b448c0a9f04b2008cc0f0f : SUCCESS in 1h 37m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/767cca856215422a8052cf3dd2f1eb79 : SUCCESS in 1h 50m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9a88dc38f4254a4a8c4c6242fc96eb9e : SUCCESS in 1h 39m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0c0c9f7d016a46ab99f9891c34bd7c3c : SUCCESS in 1h 22m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8212742abf7144b49fbc41ab57016a23 : SUCCESS in 1h 34m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/447f5b8e95cd44c7893b7ad666c7c40f : FAILURE in 15m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"938172f249ec205f43523411f85cff650dcf3bb1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-05 14:03:03.000000000","message":"Patch Set 4: Backport-Candidate-1 Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6c8035241db91ad80df2e087d5a5af24e596395a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-05 23:00:59.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9eede76cdd87524cc2e50f68967b91c060575a75","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-06 10:04:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4a1de290676440992a6fcc6e784f503af4dcce10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-06 10:04:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7b373d6704f2242e75ade2bf1a9058a51c3ff74a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-06 10:13:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4ac60f7632d9eb2e8e11f63cd71eaed6c1b75fe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 12:57:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c766373c51c45ea9d0b073cb9b35cab : SUCCESS in 15m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0f1a91a906f4f55a085fb4ef1e3b965 : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1505c094928648e3821dce742001c655 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4028d842fc934f09a12fd01c20fa0382 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cff79761558a4c918e8fb284e3a1be2b : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4cc2f5053a1f4bb78b2861fe0afeeccd : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85883137c38e49a8a5beb763cd2e9be2 : SUCCESS in 12m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9ff0c46737b427dbd276eb11ee90158 : SUCCESS in 4m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/097fc84fdd934e479493e5cc51766ff6 : SUCCESS in 6m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9db995b966ed4d8fb87a96a59537395d : SUCCESS in 9m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3c99709cefbe4fcabe8dd13a7e39d321 : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d74bb602fb124cd3bfc859f6bd1ed295 : SUCCESS in 8m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c2c060efb33407e89218b3385f152bc : SUCCESS in 2h 20m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a2b9d88e9c1744a1b42c8020ed03c856 : SUCCESS in 2h 01m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5cb5993fbe80423cb2468ffe999c144b : SUCCESS in 39m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/173f9f4a2be14e7c80234b9255a4c270 : SUCCESS in 1h 11m 34s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/dcdc13159ba548b7ae091f5dc9f5d75a : FAILURE in 19m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a3bd7fdb103b4750a101449fb58ea903 : SUCCESS in 34m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/07fe2234331f40f782ea283aba809ed3 : SUCCESS in 1h 31m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4b4cce640cf747deac28c206411490b1 : SUCCESS in 2h 40m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0ce1171563ab4f8e996dfa74b725dc80 : SUCCESS in 1h 37m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b7596b147ef646dca86b7278cee0c05c : SUCCESS in 1h 30m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/63397159b7b14774a6441ba87eb500e3 : SUCCESS in 1h 22m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/a74f326795f04ae7ac05e494baf72975 : FAILURE in 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"aa8e767d5ea2987474e29909958df5acf232b329","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-06 16:52:03.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"afe6975662b336c039fcc34f5e3ea5d5b2ff6b59","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-17 09:40:49.000000000","message":"Patch Set 6: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"064e987c20ffbed1cffcd94c7e74779b33d97831","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 09:41:33.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3ca80bb62781cc9d54af46a552f45dcf19313786","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 11:58:36.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1375b466087a40cb90b7a62a2a8e98d0 : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a170963735284e98a49949207f71799b : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/051d9a4dc8a249a2959342fe02e9aa4e : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d1698bcccb94b9ea360e3a6a502847f : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656e772e50e24431b2f47a9f6603fb5d : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5777e85a32df4ea3a41a2151dfdfd7ba : SUCCESS in 5m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/934cd34ecdec441f998679c5872ec43e : SUCCESS in 5m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/be36ceed814f4616bf9ddade96146bf5 : SUCCESS in 8m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/163673cf48ac46ec8f41d0f4cb11531b : SUCCESS in 2h 11m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be4c7b541e334f32a55de71f0b5a6479 : SUCCESS in 1h 55m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9336a94240604d5ba40af54bf13ff5c0 : SUCCESS in 38m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7c1ee1d2d3674a4b874bcb304710a91b : SUCCESS in 1h 16m 23s","accounts_in_message":[],"_revision_number":6},{"id":"c2ed0dac5d541ac7aa9787c22a030aa641bade17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 11:58:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"bea7e035120cb5f4ee86274c0206e9bd3d9efe13","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 11:59:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1da984756834981a30c796a8c5bcb28 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed897f2493484632ab83a893ecbf7d69 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"829e44acbd3b1d193ad0125c67b83ec5860a470f","revisions":{"29244385dc5fc642f845d3ba411ec429d15641e4":{"kind":"REWORK","_number":1,"created":"2022-04-29 12:20:05.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/37/839937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/839937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/839937/1"}}},"commit":{"parents":[{"commit":"2376f68bbaa48047167ce2167b19f3bf3b493e03","subject":"Merge \"Improve documentation about log offloading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2376f68bbaa48047167ce2167b19f3bf3b493e03"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-29 12:18:40.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-29 12:18:40.000000000","tz":120},"subject":"Make amphora timezone configurable","message":"Make amphora timezone configurable\n\nThe new amp_timezone option allows to adjust the timezone of\namphora using cloud-init.\n\nChange-Id: I2f49cbb7f8d99ba2da878bbfc7081a3cc3b3aa07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29244385dc5fc642f845d3ba411ec429d15641e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29244385dc5fc642f845d3ba411ec429d15641e4"}]},"branch":"refs/heads/master"},"5a102cfde5dfa5990ec081c981c561f6b46206c2":{"kind":"REWORK","_number":2,"created":"2022-04-29 12:55:50.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/37/839937/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/839937/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/839937/2"}}},"commit":{"parents":[{"commit":"2376f68bbaa48047167ce2167b19f3bf3b493e03","subject":"Merge \"Improve documentation about log offloading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2376f68bbaa48047167ce2167b19f3bf3b493e03"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-29 12:18:40.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-29 12:55:12.000000000","tz":120},"subject":"Make amphora timezone configurable","message":"Make amphora timezone configurable\n\nThe new amp_timezone option allows to adjust the timezone of\namphora using cloud-init.\n\nChange-Id: I2f49cbb7f8d99ba2da878bbfc7081a3cc3b3aa07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a102cfde5dfa5990ec081c981c561f6b46206c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a102cfde5dfa5990ec081c981c561f6b46206c2"}]},"branch":"refs/heads/master"},"924c14030c3d160d8815ce516656a17b6788d998":{"kind":"REWORK","_number":3,"created":"2022-05-02 09:47:52.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/37/839937/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/839937/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/839937/3"}}},"commit":{"parents":[{"commit":"2376f68bbaa48047167ce2167b19f3bf3b493e03","subject":"Merge \"Improve documentation about log offloading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2376f68bbaa48047167ce2167b19f3bf3b493e03"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-29 12:18:40.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-02 09:47:44.000000000","tz":120},"subject":"Make amphora timezone configurable","message":"Make amphora timezone configurable\n\nThe new amp_timezone option allows to adjust the timezone of\namphora using cloud-init.\n\nChange-Id: I2f49cbb7f8d99ba2da878bbfc7081a3cc3b3aa07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/924c14030c3d160d8815ce516656a17b6788d998"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/924c14030c3d160d8815ce516656a17b6788d998"}]},"branch":"refs/heads/master"},"402e10d902e741442c9c4f19342acb9a43a72d1d":{"kind":"REWORK","_number":4,"created":"2022-05-02 11:12:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/37/839937/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/839937/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/839937/4"}}},"commit":{"parents":[{"commit":"2376f68bbaa48047167ce2167b19f3bf3b493e03","subject":"Merge \"Improve documentation about log offloading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2376f68bbaa48047167ce2167b19f3bf3b493e03"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-29 12:18:40.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-02 11:12:20.000000000","tz":120},"subject":"Make amphora timezone configurable","message":"Make amphora timezone configurable\n\nThe new amp_timezone option allows to adjust the timezone of\namphora using cloud-init.\n\nChange-Id: I2f49cbb7f8d99ba2da878bbfc7081a3cc3b3aa07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/402e10d902e741442c9c4f19342acb9a43a72d1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/402e10d902e741442c9c4f19342acb9a43a72d1d"}]},"branch":"refs/heads/master"},"265bec28e28f4f96ad210050aa9d34aa227bb068":{"kind":"REWORK","_number":5,"created":"2022-05-06 10:04:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/37/839937/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/839937/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/839937/5"}}},"commit":{"parents":[{"commit":"2376f68bbaa48047167ce2167b19f3bf3b493e03","subject":"Merge \"Improve documentation about log offloading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2376f68bbaa48047167ce2167b19f3bf3b493e03"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-29 12:18:40.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-06 10:04:23.000000000","tz":120},"subject":"Make amphora timezone configurable","message":"Make amphora timezone configurable\n\nThe new amp_timezone option allows to adjust the timezone of\namphora using cloud-init.\n\nChange-Id: I2f49cbb7f8d99ba2da878bbfc7081a3cc3b3aa07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/265bec28e28f4f96ad210050aa9d34aa227bb068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/265bec28e28f4f96ad210050aa9d34aa227bb068"}]},"branch":"refs/heads/master"},"829e44acbd3b1d193ad0125c67b83ec5860a470f":{"kind":"REWORK","_number":6,"created":"2022-05-06 10:13:55.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/37/839937/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/839937/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/839937/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/839937/6"}}},"commit":{"parents":[{"commit":"2376f68bbaa48047167ce2167b19f3bf3b493e03","subject":"Merge \"Improve documentation about log offloading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2376f68bbaa48047167ce2167b19f3bf3b493e03"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-29 12:18:40.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-06 10:13:41.000000000","tz":120},"subject":"Make amphora timezone configurable","message":"Make amphora timezone configurable\n\nThe new amp_timezone option allows to adjust the timezone of\namphora using cloud-init.\n\nChange-Id: I2f49cbb7f8d99ba2da878bbfc7081a3cc3b3aa07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/829e44acbd3b1d193ad0125c67b83ec5860a470f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/829e44acbd3b1d193ad0125c67b83ec5860a470f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
