)]}'
{"id":"openstack%2Foctavia~840108","triplet_id":"openstack%2Foctavia~master~Ib929fbe28241bd6d3c3bd352b2d900289efd6d49","project":"openstack/octavia","branch":"master","topic":"drop-l-c","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-07-07 08:33:36.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-07-07 06:31:20.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Ib929fbe28241bd6d3c3bd352b2d900289efd6d49","subject":"Drop lower-constraints.txt and its testing","status":"MERGED","created":"2022-05-01 01:47:16.000000000","updated":"2022-07-07 08:35:41.000000000","submitted":"2022-07-07 08:33:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840108-drop-l-c","meta_rev_id":"3cc3dd14d84b933f6808c27a92522718a632fa4b","_number":840108,"virtual_id_number":840108,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-07 08:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-07 06:31:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-07-06 16:48:06.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-07 06:31:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-01 04:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-13 09:55:44.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-07-06 09:51:29.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-07-06 16:08:33.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"adb4173cc1efab87bc0ec5fe722bc035d13e62a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-01 01:47:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e84cb283e0d356f6b1335a933200f0aab4b7c5ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-01 04:10:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/57f2b3a6af4f4298a6bf78d504569025 : SUCCESS in 6m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1445c71bc6e24bd788d77dfe8db39f5b : SUCCESS in 2m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e82213e3754492db40f388a80d20689 : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f0728cac68405788efea8e5c7168f2 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5beef9b8b1c4eb193689e05c98d364e : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b90f45b66924630a610a9a7901b575d : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8abe66a234da496db9e2e67ca3b31d8c : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1611e94abe184de69db7e0890543aba8 : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06fb732d016f4549811f3bc5b74c5790 : SUCCESS in 4m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4f280780381b43f19dd2708ad34e0ae4 : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/85f560d47a0c4eefac62d067a4b6a3b2 : SUCCESS in 8m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6ca20c07f68d4927bbc92dd518b7aabf : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c4fd8f820adc4243be4b03bc5f2e50d6 : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4f9449ad8003495eb245cf252d9a5bff : SUCCESS in 2h 16m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8df5313ba5a24d07a20f994573006f06 : SUCCESS in 1h 51m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eb10a094ac8a405f8c04bdc1cd794e86 : SUCCESS in 44m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dc1e724ba7a544b096b9264a273f0203 : SUCCESS in 1h 08m 37s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/78dac9ff40bc491bb1c40685851fb371 : FAILURE in 19m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/39ce4d1df0ff461591de62b5a3f7836d : SUCCESS in 34m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a1195cdc1fc04a93a6ed31fac9fdcbdd : SUCCESS in 1h 58m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1a9af002264740cc95ee96dc1f70890d : SUCCESS in 2h 13m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8cdbda4c7cad46fb98fe9ebbfb95e526 : SUCCESS in 1h 48m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cb72422b01594014b25f17d485bc55eb : SUCCESS in 1h 37m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0e3013dfc3604b0d8f47aa767390ccdc : SUCCESS in 2h 06m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/826af2627bab494cab9d638a809984b6 : FAILURE in 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bec6a0a523339328e36ffc14298f78fc8ae7de95","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-13 09:55:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5b1aff164feb280660334ec33c2b5d115537f1d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-06-21 20:10:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f5725a8e49358711aa87c968ea2fe5f1b9faa327","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 22:14:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ba62292c20eb4be78858e09324801f11 : SUCCESS in 5m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/be6769ecf3314f37a86a2cc9f3c3beb2 : SUCCESS in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5b57f16591642768d5a2bef6c554fff : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b3b438cefc44065b2ff0d8f52bec538 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/deb790d67f1a46e6bd156efd9284ea50 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ced231e66a5f49e3964ff39d35bec37b : SUCCESS in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7e52ce7eaba4b06bc3c3100e6c825dc : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b7bfc94a6143c2a3ca3a046ed660a3 : SUCCESS in 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00a9fdf3ff6a457bbc603decddc84944 : SUCCESS in 6m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/cdac183dffde4aa99157e3646105af1a : SUCCESS in 6m 50s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3208115f39324ca6aba8dbf6852e7eaa : SUCCESS in 9m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6e9011ddfe754394ab4d1adb5f328f01 : SUCCESS in 5m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/05418a96c9174a84b5e4460ff1c1f703 : SUCCESS in 8m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/10a767ad36874d68b5a74034a8d101bf : SUCCESS in 9m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/23aac0e9bcd84543852c67594ed46044 : SUCCESS in 9m 39s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/87e5d6928a3a4242824329a75931bfbf : SUCCESS in 1h 55m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d83722c879b44b618d43cab662032336 : SUCCESS in 1h 35m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c8c2ee6f126d47e38775e3817e9e8a0e : SUCCESS in 34m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6dbc775002674c88a3a50d0ee8f49701 : POST_FAILURE in 1h 11m 57s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/797478a4788048e5ab053a643251da7c : RETRY_LIMIT in 10m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f4afb761314439ba3fa9a3cad742e6f : SUCCESS in 31m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/918afe9c25904733a8ebec6ae621ca43 : SUCCESS in 1h 31m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2a8f4327866f4545a221103f9877946c : SUCCESS in 1h 59m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/da39cbacc1824baa83f1a4ae3da55b5f : SUCCESS in 1h 29m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a26a4c32e3d14928b5ed1711627234fd : SUCCESS in 1h 30m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/aff0564d7a2747088e9127ecf14f34de : SUCCESS in 1h 17m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/31fafc76229c4ea39ae4daa9922baf40 : FAILURE in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"538c4df9a336cabe9a4feadd6470298064233a80","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-06 09:14:08.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ee1c73d7b70eeeaeaad06d84021a41ad866ae98","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-07-06 09:51:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c38aff4ff48a82fdf60ba53ef8eb8f7714836a5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 11:09:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f67b1e7df71942589808cce9be41a0c0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/55bab9c6fdd0432eb09983f86050d5af : SUCCESS in 4m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/55cfee62b6644747bd318cac648ca809 : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76670ea4f9c34669a9834768700409f7 : FAILURE in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8464bcdc318048af8c9056c93f40a31f : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a69b391922f4d2886bf472c5b0eba9a : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef872df947c748fa84c0380ec196124a : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96d0ee2c5e6748ddb97253b7a5b7efee : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf0ec47fe67c488082f8d759dafe8155 : SUCCESS in 12m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c69eadede994e71ba342a0e15734d8f : SUCCESS in 6m 04s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7253e3108d614275a4c1935d74c6bc9f : SUCCESS in 5m 32s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3800b1d351184701933de11d59824f60 : FAILURE in 6m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/85b8c76bafc84bedbc9855c8034fb107 : SUCCESS in 4m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a90a0b6538e040f39d430467a94ce8a1 : FAILURE in 5m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cebd2252c91843709325e336d461d11c : FAILURE in 5m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/82cafd20bafb478eb7b8bb1994ddc595 : FAILURE in 5m 59s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/97a4a1ef01e24a4b8550b372c0affcc6 : SUCCESS in 1h 52m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c9f9a44b7f554d07a32249882b5c7365 : SUCCESS in 1h 26m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b4fd5410ba124b41a77a9cef71fd3f33 : SUCCESS in 37m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dfc7253824b24deeb5a6120e14384b73 : SUCCESS in 50m 10s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7e110c9787df4cdf982addf08893d727 : RETRY_LIMIT in 8m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec5afa2189fd4a9aa941214ddfeac8d8 : SUCCESS in 30m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/695d6808f2114292b362f29c93b9fea8 : SUCCESS in 1h 37m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a5b006acacdc4b6094aa9096c8568ace : SUCCESS in 1h 34m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5baf6b029e3d453daf11339f70a66508 : SUCCESS in 1h 50m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4fee3b1295284af8b65c5ee727137bb3 : SUCCESS in 1h 38m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/98fc5e2eb823419f99a2b6dee1888c34 : SUCCESS in 1h 39m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/f8b9f929fda54dc789bf21960a6aa9bf : FAILURE in 11m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f70ef6e620e811de4b39fc46dfa0dd3b4519c1a7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-06 16:08:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d4678bb833dcd887f686201c9bdff66dc1e42932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-06 16:44:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bf8480c304a0dd228b5337dff3033d645c18f383","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-06 16:48:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"44692d38190e2ec0d6ea50d0799424dc939d85bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 19:06:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/752c058434684b0c97ee14ab6c10ca9a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2303983087284c7d8a622cb1eb58a3d9 : SUCCESS in 5m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/10c2ab202f544318b25d72c538db925e : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/542f8f06fbb142ec874397cb0f88ff44 : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c6214fca9664ca6b188ab0b58b65f79 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fde84de392734806a20dd1cb073b7359 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3dfcfbd980114b8f8522596333867fd3 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/254ee5c2bf0a4c13bf60bc6a58a9b8fa : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34b557f12ed746179f1a8427d7b51319 : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5aef0668b8ad4318b17dbd591965ff27 : SUCCESS in 5m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/405c6e5745cb421d899b28183a0f247c : SUCCESS in 5m 17s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/56ace721c160427d94b94038e4bf2ceb : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fb8bcb31f4424e11ae76900f631216c5 : SUCCESS in 5m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cf89bf78fd2f4c7182b6af88c739fd6b : SUCCESS in 7m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/089ab99d1aee44a7acca6c96b040edc1 : SUCCESS in 8m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9f562f81894243fa9ab26db69cf98cd5 : SUCCESS in 8m 42s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d28c75b5155a4b8ab00b72431731abda : SUCCESS in 2h 01m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3a964ac50694669b7dd15ae68bc9bfa : SUCCESS in 1h 56m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9865df3b1a0d43c38147509e8c9c261b : SUCCESS in 41m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a7d6de3460a743faa5955df248fc2b73 : SUCCESS in 1h 12m 28s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/67df316769034c18bad040258d3fa9e5 : RETRY_LIMIT in 12m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce3e05f0c6254c30926814624efe1fd1 : SUCCESS in 39m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/35c7e8e6ccb6487b9799a98b613aac0c : SUCCESS in 2h 06m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ab6bea661f0a41cbad9709b8a0d35624 : SUCCESS in 2h 18m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7b3b4cbe54fb4954847fb73c2fe0cb70 : SUCCESS in 2h 11m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fbfa9b31152a41faa8892d2cfeecdec4 : SUCCESS in 2h 09m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/662faaf3c40c4a549f121191188dd4e0 : SUCCESS in 1h 26m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/f7177b4167cd4c30b047f1264d9640d8 : FAILURE in 20m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"676cb3dcd058c6c4bdd6e51a0b2bf8a813c6f16e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-07 06:31:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"af455debc575a37e4d7279e9b82f7babb63d6673","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 06:32:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ab2e4508ebe7b952082936f6d547a8fbf89a8ed2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 08:33:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/461d3ee0ee7341f18b9df81d076ab891\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cb44518740b749d18f4b3edcd6903b2e : SUCCESS in 5m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d73598f1afa841359c9a5e01a5f48ce0 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/395e44a2879d48dd844b9ff7a5e5361f : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/864f8a0cf0f14c2d9e55236b2639285c : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33f774fdb26c4034b63dc1c38d383abd : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f29af1e8453b46abaa3143f278e0c07c : SUCCESS in 11m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1307fdf67edd4d7cbec3945e5c51983a : SUCCESS in 4m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8dcdac7b533e469c9c26b3b0e38318da : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1ee10d12b91541f8b4119a255e48b3d1 : SUCCESS in 8m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c3326a8c951645929a2b5c70a1b8245a : SUCCESS in 1h 49m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ef80d0060354bf5a5dab3a5eff67713 : SUCCESS in 1h 40m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bc1cf3e0337d49a29c5922fbfdfbeb48 : SUCCESS in 36m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/82c3a6cd989e4e66892433636286323c : SUCCESS in 1h 07m 39s","accounts_in_message":[],"_revision_number":3},{"id":"fb2633c5ee76f3857a3871edc8936d6b5fc33337","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 08:33:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3cc3dd14d84b933f6808c27a92522718a632fa4b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 08:35:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4c134c4fa6c4876913e06a2f0d4fcc3\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/d320d0693aaf4577af3fcfd35b46d28d : SUCCESS in 45s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed222697b85147ce84a994998e74e3d7 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8660c36586a14b00ab83a0623482ad96 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"38cb7b6fe8e52eb29efdc8cbc08619af4ab7d4c1","revisions":{"dfe9f83e6be34d163b895ede66dfefb5584d74f0":{"kind":"REWORK","_number":1,"created":"2022-05-01 01:47:16.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/08/840108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/840108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/840108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/840108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/840108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/840108/1"}}},"commit":{"parents":[{"commit":"5d7bafb42c482bab545a6d023ac411657f1cb912","subject":"Merge \"Fix potential race conditions on update requests in the v2 worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d7bafb42c482bab545a6d023ac411657f1cb912"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-01 01:45:48.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-01 01:47:15.000000000","tz":-300},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: Ib929fbe28241bd6d3c3bd352b2d900289efd6d49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dfe9f83e6be34d163b895ede66dfefb5584d74f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dfe9f83e6be34d163b895ede66dfefb5584d74f0"}]},"branch":"refs/heads/master"},"688f93c471f8b7c328508fd4ad5fc508ed639c81":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-06-21 20:10:32.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/08/840108/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/840108/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/840108/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/840108/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/840108/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/840108/2"}}},"commit":{"parents":[{"commit":"f166bb1481fb58acd7e7329ec394f43249001718","subject":"Fix duplicate object error messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f166bb1481fb58acd7e7329ec394f43249001718"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-01 01:45:48.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-06-21 20:10:32.000000000","tz":0},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: Ib929fbe28241bd6d3c3bd352b2d900289efd6d49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/688f93c471f8b7c328508fd4ad5fc508ed639c81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/688f93c471f8b7c328508fd4ad5fc508ed639c81"}]},"branch":"refs/heads/master","description":"Rebase"},"38cb7b6fe8e52eb29efdc8cbc08619af4ab7d4c1":{"kind":"REWORK","_number":3,"created":"2022-07-06 16:44:51.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/08/840108/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/840108/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/840108/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/840108/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/840108/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/840108/3"}}},"commit":{"parents":[{"commit":"f45388643f0423199c9ac86a21a163e535cb0007","subject":"Add WebTest as an indirect test dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f45388643f0423199c9ac86a21a163e535cb0007"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-07-06 16:44:21.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-07-06 16:44:21.000000000","tz":120},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: Ib929fbe28241bd6d3c3bd352b2d900289efd6d49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38cb7b6fe8e52eb29efdc8cbc08619af4ab7d4c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38cb7b6fe8e52eb29efdc8cbc08619af4ab7d4c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
