)]}'
{"id":"openstack%2Foctavia~840182","triplet_id":"openstack%2Foctavia~master~I63474acbcea5106ea702d21cb238aa57cef5d96d","project":"openstack/octavia","branch":"master","topic":"needless-sudo-in-lvs-masquerade-sh","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-05-05 10:39:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I63474acbcea5106ea702d21cb238aa57cef5d96d","subject":"Remove unneeded sudo in lvs-masquerade.sh","status":"MERGED","created":"2022-05-02 13:20:49.000000000","updated":"2022-05-05 10:41:38.000000000","submitted":"2022-05-05 10:39:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840182-needless-sudo-in-lvs-masquerade-sh","meta_rev_id":"fc3991f4727f2a0bba5434f4faf569f2ef2a0a2b","_number":840182,"virtual_id_number":840182,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-05 10:39:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-04 17:22:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-05-05 08:35:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-05-05 08:35:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-05-05 08:35:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-02 16:53:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-04 17:22:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-05-05 08:35:30.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"a5640e0cb4fded813bf233c65a184589ab96a404","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-02 13:20:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4db1f577b71c7865f3dd5acc1d13a815a6bc0ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-02 14:15:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1dd4212ad86588e0675344594c70ed7b611f2a45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 16:53:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e71189a4e204c4584089fe48bc8fad4 : SUCCESS in 10m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1b1c83e3fb540b08a92e21c1dbc2096 : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/754c40c0a4e84beb86463f8f4f78516b : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/affaa01035f246fe87ebbd30cc5197e3 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/753709ea7bd247149747170a9480a286 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47c34535005149389663c5b8115fc2f8 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cad570ad8f145f198386918a7cab308 : SUCCESS in 11m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9d2dbcfb58f64fa6a343f315486be106 : SUCCESS in 6m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/49cdd9674c364469ac7dc79550da201d : SUCCESS in 8m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ec08b2cb00f7448f8f5aaa0774e9e796 : SUCCESS in 5m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/25e39916c6d24b8ab75a35ef45ad00df : SUCCESS in 9m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c7a6746f97541f58fc63d56d402e3c1 : SUCCESS in 2h 12m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dbb450d8fbd44600a6ae0be6fdd678e2 : SUCCESS in 1h 34m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2bb7b37f959a47d5a5f033b1600673f6 : SUCCESS in 35m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2c658d9d13e24b1e9aaa24b9452a6ad1 : SUCCESS in 1h 39m 26s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/441cda1564c84cec98aa7955195252bc : FAILURE in 44m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20bc699b92a24a759a60960c17de8960 : SUCCESS in 31m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e39262af761e4fb28a764b1be2b49bde : TIMED_OUT in 2h 35m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9176a90761194a9daf25e63bdc5330ee : SUCCESS in 1h 45m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/97cabf9c29164bc9a7b13e4b2de0396a : SUCCESS in 1h 49m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e5f3e99bbd384366a46c769317a64c97 : SUCCESS in 1h 34m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9c58979baba340849107e90c462b7727 : SUCCESS in 2h 29m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/0261931878e64b0e9465c81241b0ef8f : FAILURE in 39m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"343b25ff3fa9dbbd237c135291e54aa0e6e6ac07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-03 05:57:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5bcae0ac029717bccbd6bf6813ce52912e7460db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 07:47:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4b06867c8204b7f82fe68fd7a413f0b : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e93b795df354113a0de624b4bfd492f : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5030e4d5988f4ee1956d15b6a7019e99 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/068bf5a8a6e64a3e887f491285ffb846 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c165ab29056470f941a27732349444e : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba8207000e664bc48aa1ba20ccf9333a : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd2b416742c942f6af0a26aaa39a64d7 : SUCCESS in 9m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcdd8f2f727e4c68948fe800dedbc192 : SUCCESS in 5m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ec71e659f441451fa160eb3156ced789 : SUCCESS in 6m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a51337c8206447eb8295cd1b23f9aa9f : SUCCESS in 9m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d7469b4f4a934b6492ee6d4096afae8b : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b17de772eb06463b83f242d914c1f636 : SUCCESS in 7m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d4bb739e914b43718af66f3b0aa3347d : SUCCESS in 1h 33m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aef75a41048c434d91cf9ec65fa71f5d : SUCCESS in 1h 28m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0078855728414683b15b754f0bee36c3 : SUCCESS in 34m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0ac6c2c6a96f4cf7b01bbe38b3afc6af : SUCCESS in 52m 50s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d0161291dd6f4ede91d0d7a0a959ad33 : FAILURE in 14m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d41b1c205d7a43a6a5ca6d52261c7a91 : SUCCESS in 35m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/382f593dfd2943e38b8191b5fd8f4241 : SUCCESS in 1h 39m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8ee2c5f99bdf45bfb64e8e13e48a2653 : SUCCESS in 1h 48m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/346e450ff63b4113b1dcecb804a1b2b5 : SUCCESS in 1h 37m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/02989fb41a834886b57a01d513b4dc26 : SUCCESS in 1h 30m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/29d56c52359046f0831e8da77ffefc2e : SUCCESS in 1h 36m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/4eb4980a9149459db60178d77c5b3d30 : FAILURE in 12m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"05291ebf3dc63cd6a330e4bfd30a6574e4d5f415","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-04 17:22:49.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7c9b43df2b4b79c690d2b5096c4dc8752388b4a6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-05 08:35:30.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"82bc457de1f050f36f8f5956d4b456048ac1f99d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 08:36:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ac0a277c107275bb8e1daf4be9aaf8cd2cd534f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 10:39:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fc95578197b4bf8bf774e32934816d7 : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a99a5a12465489ebf4e6a0506649853 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d2f1537d59a4cb2992a59c49acfd091 : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/465ba5b3c64b4013b5a3278a6b00d1e0 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7de6eddd0db4e8290c86f4828493cc8 : SUCCESS in 11m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcb1128f9a93483186d337be02eb8d2e : SUCCESS in 4m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7767184b65b542c19a5dbadc12b750cb : SUCCESS in 5m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1fc3e9b2d43b4544bf0ba866459ae0e6 : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d84f750f3a094ed3890ed31d2843792a : SUCCESS in 2h 02m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59526ac7a1c043cf81bcd09774e91527 : SUCCESS in 1h 30m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1be707d8a55644d182ca979ba7863e46 : SUCCESS in 36m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4e981c1d700e461ab763bae27e86877d : SUCCESS in 51m 51s","accounts_in_message":[],"_revision_number":3},{"id":"c24c3ee8a451e57b8c292d0005a83c683b9ac431","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 10:39:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"fc3991f4727f2a0bba5434f4faf569f2ef2a0a2b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 10:41:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a248fd3e74945dcbc4f911a6aa45e20 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28f1c3692d8040b5b3c41d9a0631ecba : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"72bdc0f88d2b9fa6fdf5ef879cb977d45ca0cc2e","revisions":{"1f3dd22005eab53202614ac756dfc8377e0e6643":{"kind":"REWORK","_number":1,"created":"2022-05-02 13:20:49.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/82/840182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/840182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/840182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/840182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/840182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/840182/1"}}},"commit":{"parents":[{"commit":"2376f68bbaa48047167ce2167b19f3bf3b493e03","subject":"Merge \"Improve documentation about log offloading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2376f68bbaa48047167ce2167b19f3bf3b493e03"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-02 13:20:40.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-02 13:20:40.000000000","tz":120},"subject":"WIP Remove unneeded sudo in lvs-masquerade.sh","message":"WIP Remove unneeded sudo in lvs-masquerade.sh\n\nChange-Id: I63474acbcea5106ea702d21cb238aa57cef5d96d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f3dd22005eab53202614ac756dfc8377e0e6643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f3dd22005eab53202614ac756dfc8377e0e6643"}]},"branch":"refs/heads/master"},"414319f6961ffa61a71b8027d30df68828fba92c":{"kind":"REWORK","_number":2,"created":"2022-05-02 14:15:44.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/82/840182/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/840182/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/840182/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/840182/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/840182/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/840182/2"}}},"commit":{"parents":[{"commit":"2376f68bbaa48047167ce2167b19f3bf3b493e03","subject":"Merge \"Improve documentation about log offloading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2376f68bbaa48047167ce2167b19f3bf3b493e03"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-02 13:20:40.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-02 14:15:36.000000000","tz":120},"subject":"WIP Remove unneeded sudo in lvs-masquerade.sh","message":"WIP Remove unneeded sudo in lvs-masquerade.sh\n\nChange-Id: I63474acbcea5106ea702d21cb238aa57cef5d96d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/414319f6961ffa61a71b8027d30df68828fba92c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/414319f6961ffa61a71b8027d30df68828fba92c"}]},"branch":"refs/heads/master"},"72bdc0f88d2b9fa6fdf5ef879cb977d45ca0cc2e":{"kind":"REWORK","_number":3,"created":"2022-05-03 05:57:19.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/82/840182/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/840182/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/840182/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/840182/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/840182/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/840182/3"}}},"commit":{"parents":[{"commit":"2376f68bbaa48047167ce2167b19f3bf3b493e03","subject":"Merge \"Improve documentation about log offloading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2376f68bbaa48047167ce2167b19f3bf3b493e03"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-02 13:20:40.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-03 05:57:11.000000000","tz":120},"subject":"Remove unneeded sudo in lvs-masquerade.sh","message":"Remove unneeded sudo in lvs-masquerade.sh\n\nFixes issue with SELinux and the lvs-masquerade.sh script on the\namphora.\nThe script already runs with root permissions, so the use of sudo\ninside the script is unneeded.\n\nChange-Id: I63474acbcea5106ea702d21cb238aa57cef5d96d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72bdc0f88d2b9fa6fdf5ef879cb977d45ca0cc2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72bdc0f88d2b9fa6fdf5ef879cb977d45ca0cc2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
