)]}'
{"id":"openstack%2Foctavia~840643","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I63474acbcea5106ea702d21cb238aa57cef5d96d","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-05-07 10:43:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I63474acbcea5106ea702d21cb238aa57cef5d96d","subject":"Remove unneeded sudo in lvs-masquerade.sh","status":"MERGED","created":"2022-05-05 10:36:30.000000000","updated":"2022-05-07 10:46:46.000000000","submitted":"2022-05-07 10:43:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840643","meta_rev_id":"00e5a37ca5b1ef3d16756c6c45bfec801596d559","_number":840643,"virtual_id_number":840643,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-07 10:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-05-07 08:05:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-05-05 16:41:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-05-07 08:05:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 13:26:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-05 16:41:40.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-05-07 08:05:39.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"5f5faca945a63321718134e0ae66f9d5210779f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-05 10:36:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce96bd0d23c6bfc659c0929dfb8aa4e3b4cb2df3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 13:26:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ac7fef29dd8434d9ad7fb37bac79014 : SUCCESS in 11m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e172715bbb5141ba94ab747949ccc478 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e11402eded924a56acd4aa8990756960 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f65b5f9953b4bcd9debb021b298a988 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa60426a6ece470e961eb67598897dbb : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5f0611925434a82a96b35dadb29d8a8 : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8ff6639e9b74b2ab3809daf0f8efe36 : SUCCESS in 5m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7c98f0180d6a437a9add55123018fc93 : SUCCESS in 7m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/01b199648bd248d4ae05f44e3257b751 : SUCCESS in 11m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7f6ed16c627d4f05a938c933ecc0e4a2 : SUCCESS in 6m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8a24a8fcab5a45258dbf85666cc2c026 : SUCCESS in 8m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bbe519fb303a4a2796db5fb29c5ef6cf : SUCCESS in 2h 02m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46ca00de04894ab49d117908f92df9d0 : SUCCESS in 1h 32m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/895510fa99c7488fa53a4bbdae84b9d1 : SUCCESS in 33m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/828b5989313042e5bd8df7939a34c5b0 : SUCCESS in 48m 52s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e624ec7440e44b6eae9f522a88ecc702 : SUCCESS in 1h 03m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc1dc38f33ca41fc887b499e19be162e : SUCCESS in 33m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a13f762b57ca49ef90fc0e813ca9f178 : SUCCESS in 1h 26m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8b4fbbf59f9142fbbf920ae0c116ad3c : SUCCESS in 2h 44m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f6cbbd7167154fe9b76bc9c85e09910e : SUCCESS in 2h 08m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eddaf6bcb63543a983ea81c0ca2c64e7 : SUCCESS in 1h 34m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3764aa6bbaf74fd6bed2b83e73bab753 : SUCCESS in 2h 04m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/4de3344e86d84704ba0115a0951d4906 : SUCCESS in 2h 12m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a92e5eb26fe60c889c9f2df86aff9b3291fc7410","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-05 16:41:40.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"483c4b15eb2ce61b08180ab06cfbefdf8552d496","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-05-07 08:05:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1eaf1f7923e68ea03f32d4bcb34f2429f05cabfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 08:06:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"749e6c0d49304d2ea89a91d714ec77bca8b5c0fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 10:43:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2000853986274b75bdab352579dd900b : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8f921fca867400884b6d67afb7c0869 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42586ec4daaf4c57858fee5d3c10a7ed : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62cb6ef9651048aea6916b9576f28a70 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1615cea024974ed58dc171203e967fef : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5ca8b1d143542da8bdb1d426a5a5a0e : SUCCESS in 4m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e28411d0f00b4aee86c51b2826f0c01b : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b2ce8ab55a164c5aa4ef04cd67c56afb : SUCCESS in 8m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/486ab7bc8d2b47a4b743876fd5c72acf : SUCCESS in 2h 30m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31b0e327c98b472b9526f422f1d99a9c : SUCCESS in 1h 29m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0c8061a2fd7d4d2395e52d0237b2762b : SUCCESS in 33m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/40f7193994ec4d51b352dcf8906e541f : SUCCESS in 44m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8eb73b590f575b33bb84c4354807aaf1bd66dbd0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 10:43:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"00e5a37ca5b1ef3d16756c6c45bfec801596d559","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 10:46:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9ee571557964cef833f4fe82cc49a09 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58a1c4c5c72c4608aa9331dfe60e3d99 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0df7c6b7402ac370de410aecb91c064b3805e8c2","revisions":{"0df7c6b7402ac370de410aecb91c064b3805e8c2":{"kind":"REWORK","_number":1,"created":"2022-05-05 10:36:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/43/840643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/840643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/840643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/840643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/840643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/840643/1"}}},"commit":{"parents":[{"commit":"7aaac224cb4296a4802e4c21c76ba9d010335333","subject":"Fix AttributeError in exception handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7aaac224cb4296a4802e4c21c76ba9d010335333"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-02 13:20:40.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-05 10:31:17.000000000","tz":120},"subject":"Remove unneeded sudo in lvs-masquerade.sh","message":"Remove unneeded sudo in lvs-masquerade.sh\n\nFixes issue with SELinux and the lvs-masquerade.sh script on the\namphora.\nThe script already runs with root permissions, so the use of sudo\ninside the script is unneeded.\n\nChange-Id: I63474acbcea5106ea702d21cb238aa57cef5d96d\n(cherry picked from commit 72bdc0f88d2b9fa6fdf5ef879cb977d45ca0cc2e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0df7c6b7402ac370de410aecb91c064b3805e8c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0df7c6b7402ac370de410aecb91c064b3805e8c2"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
