)]}'
{"id":"openstack%2Foctavia~840694","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I0f4c59a2928f2a813171109aaf73a1d7dff9eefe","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-05-07 10:41:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0f4c59a2928f2a813171109aaf73a1d7dff9eefe","subject":"Reject invalid whitespace in HM url_path value","status":"MERGED","created":"2022-05-05 16:12:10.000000000","updated":"2022-05-07 10:45:45.000000000","submitted":"2022-05-07 10:41:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840694","meta_rev_id":"054b5381a5c4a78916d7545f56e6d3957db9303e","_number":840694,"virtual_id_number":840694,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-07 10:41:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-05-07 08:03:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-05-05 16:44:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-05-07 08:03:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 16:44:15.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-05-05 18:29:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-07 08:03:44.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"74cae32f800f39caccdc385ed0e6cea688ab06f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-05 16:12:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea4d2ba84bb264b0ed326e331327bed3bacf3c04","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-05 16:44:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee65a65ce8b1b0a3e744662b855b9a21899843d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 18:29:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaa950eafc26459384a9c15745a968ad : SUCCESS in 9m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/baebf38e5ca44130b9762a15fe4fd2e0 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ec4940ca6ae459ea715b66f1a2de8cb : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5aa984399c3345b8a18c59bfc031871b : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d86cb545511145b49df5091b91ac0928 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/985fecb2e39a443dad60540ea1b70b87 : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b64a2029f9874ffaacf731e2eb6a33c9 : SUCCESS in 5m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/933a368ecd5e45ac894aec5fb1f73a95 : SUCCESS in 6m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1c3b324401324e11af18647f1b26b1c1 : SUCCESS in 9m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c2741befc7c94707ad4764b7a5a0a292 : SUCCESS in 6m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/82bfba3681084f1b880beb54c37bcd88 : SUCCESS in 7m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/082155179fb745e5a07592221b4186cb : SUCCESS in 2h 08m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af477e5a32b746beb6d319d3c442637d : SUCCESS in 1h 40m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0adb38eb081642af9c12cdd7d6a18330 : SUCCESS in 43m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b7cdf940825c4ba69a8644b064a0fde9 : SUCCESS in 46m 34s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b41fcd480e7741ceb892e01a7927d073 : SUCCESS in 1h 05m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d58ec68219c64bffb755cdac924f5ae6 : SUCCESS in 44m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7933e6c4ed37421b9b8160d93753af00 : SUCCESS in 1h 49m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0b0b4acee4f0401788f35d77243c7ad7 : SUCCESS in 2h 13m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/20d647c61dcf462bb599d6ee13ea066e : SUCCESS in 1h 37m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/73d332349787484396c0c0418e79752c : SUCCESS in 1h 37m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/941ea3621aae451fa115f0f26c2519f4 : SUCCESS in 1h 45m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa093f298a324e11ba0406c4279e4c31 : SUCCESS in 1h 31m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d92d9da8c8e5f98d7d61c0e4ece29263899904c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-05-07 08:03:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d3ee127c85c7081c81b840a00374b662fa4f7630","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 08:04:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"369390f6bb280064d7f1a7dde3938e15676681cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 10:41:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c91f398d0b1f43d8a2f54705c2581cdd : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffea85ba7d014b3c8f01f50aa4f6e078 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0618842dd51044ff95895455f6a3b705 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc1ecae0092a44f5b58c5d08668f68c1 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae73b8e2ff4b41d483d4619e2cb714b7 : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b603ed2c9c7b4910b32127a93bd3dc4d : SUCCESS in 4m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5288d972d01249a79aee5dcbce26dad2 : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/60382f3c8a8947d088549688ffe12ac0 : SUCCESS in 8m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0a20bfa5079a4f3599af838df00b4f74 : SUCCESS in 1h 30m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/860f53cfd1c944d192ee6782d0f8d9e1 : SUCCESS in 1h 42m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6d96cf0f199b4c62b4a4b2047b904b27 : SUCCESS in 36m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d5a1444872c04222bcdced85860e6645 : SUCCESS in 47m 46s","accounts_in_message":[],"_revision_number":1},{"id":"6577b5b73839d076ea7e7d840a673d9623541ab2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 10:41:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"054b5381a5c4a78916d7545f56e6d3957db9303e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 10:45:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09516e7a5357484bad83803155c54aef : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab65598952d84cc2b753c4f86db4ff9c : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9afb3af6bb44cff6b0b0d92ed8b87c910cab4092","revisions":{"9afb3af6bb44cff6b0b0d92ed8b87c910cab4092":{"kind":"REWORK","_number":1,"created":"2022-05-05 16:12:10.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/94/840694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/840694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/840694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/840694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/840694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/840694/1"}}},"commit":{"parents":[{"commit":"c6da46ea8d1d6652f8b8eae9e3815559b9612547","subject":"Merge \"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6da46ea8d1d6652f8b8eae9e3815559b9612547"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-22 09:02:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-05 16:11:58.000000000","tz":120},"subject":"Reject invalid whitespace in HM url_path value","message":"Reject invalid whitespace in HM url_path value\n\nPlain whitespace characters in url_path must be rejected because\nthey may allow authorized users to inject arbitrary HAProxy\ndirectives in the configuration of amphorae.\n\nStory: 2008994\nTask: 42656\nChange-Id: I0f4c59a2928f2a813171109aaf73a1d7dff9eefe\n(cherry picked from commit 251fab17b51f99f1f0c1cce4985693993d8fb957)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9afb3af6bb44cff6b0b0d92ed8b87c910cab4092"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9afb3af6bb44cff6b0b0d92ed8b87c910cab4092"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
