)]}'
{"id":"openstack%2Foctavia~840696","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I0f4c59a2928f2a813171109aaf73a1d7dff9eefe","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-05-07 10:41:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0f4c59a2928f2a813171109aaf73a1d7dff9eefe","subject":"Reject invalid whitespace in HM url_path value","status":"MERGED","created":"2022-05-05 16:13:26.000000000","updated":"2022-05-07 10:45:12.000000000","submitted":"2022-05-07 10:41:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840696","meta_rev_id":"15e5b1ca51d24c1a5bfb70bbc7daefd19229344c","_number":840696,"virtual_id_number":840696,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-07 10:41:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-05-07 08:04:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-05-05 16:44:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-05-07 08:04:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 16:44:52.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-05-05 18:01:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-07 08:04:02.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"a88a55986b048cf0419db54976c2dc8cad776f3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-05 16:13:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eff601cf3cd2dc7ac3cd1095161e9e8d30711326","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-05 16:44:52.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e858dca0b8495be64aa6682c34e1d45033f08f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 18:01:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/441420ebc8104c2a863a7031463be08e : SUCCESS in 9m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7b3ca27b3a0469790a880055cc41ecc : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/707b91b413f94bbf89d15999bcc5ce61 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c75060e383ca4059854848c7ba9da580 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fd787c46fd142a5a46f18e34c6f254d : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e2daf1c8d1e40769ac807660180ef70 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b17a10f0596f40158dd055354aff6d00 : SUCCESS in 11m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/165ede5200ca42588d920b37e84afd64 : SUCCESS in 4m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/16d9e4492b5b4b4c8b546066a71301f8 : SUCCESS in 5m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/17d73a70c2e345b19d082f052ab57958 : SUCCESS in 8m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8a8681ee4464404cabfa3bc680c0a9f9 : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b6c6a498d7b74e8cb8a8411ac56c9ba5 : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f09e09539aca4741aa7399ddd525497a : SUCCESS in 1h 40m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b85b02be67d74545a6b788dd340f2801 : SUCCESS in 1h 35m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d759edc2c9954065b1e92a5ef2c91ac0 : SUCCESS in 38m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a5251a84f3934802adf5a331bfc830f8 : SUCCESS in 34m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/438c77403ca04205948a479746259c88 : SUCCESS in 35m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/aef52fa8d5f64b31a25f91ca1f284de6 : FAILURE in 40m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/77548da6e76d4ee3ba2195966387ee97 : SUCCESS in 1h 31m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8514b2f65d4c441db173cc4a52b96640 : FAILURE in 32m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/334c4b979158462b946e5f1314eec039 : SUCCESS in 1h 41m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1fbb735ff0b1484e98cdb6d61679ef44 : SUCCESS in 1h 43m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/79119e2c09b54f14b20901ad359c1a38 : SUCCESS in 1h 38m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a228f1965eb349bf99d18225c8f1b067 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d830088842b77b4aa72cda71141c7cc7a6e5f2df","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-05-07 08:04:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bf459b5da4874a6b6d1751ac7f51fc788b1c3edd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 08:04:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"830781cf1b1cbfacd0de675ad959df3c7a79f2ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 10:41:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/994ed8d10302481086328d349efa23e0 : SUCCESS in 9m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fb2144707854b74a333e9b7dce374df : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e7832a3d5144712b6a19f54764c80f9 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/731ce84e7ac3431ebf4b48b04fd81816 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a42984bee6f34d9ca42eee984a79c481 : SUCCESS in 12m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72810c4acba8485cbea47568960e3998 : SUCCESS in 4m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2231a67744be4d859078551d242c59bf : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a267ec19089547f4a0624c2083699a67 : SUCCESS in 7m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/82a33438873c4b1d9ba7d8ec1d9edfd7 : SUCCESS in 1h 45m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9076ce73d2e4638a4cbb2b9e6ee5322 : SUCCESS in 1h 23m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/86fdb1d03f25454bb784ed65d56e6839 : SUCCESS in 49m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c727534c2c394140b94e040c64c3b266 : SUCCESS in 33m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9694c106bc04d05b33586b3e4fd4101 : SUCCESS in 34m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ef4aaf31494bfbf909260547030529e439d5992a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 10:41:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"15e5b1ca51d24c1a5bfb70bbc7daefd19229344c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 10:45:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1028ba547dac41c6bba5a53c49492a64 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fe95750fe7743ccb8e565e50af282dc : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4503bd2767a33665c3474e4aa67ff45824222377","revisions":{"4503bd2767a33665c3474e4aa67ff45824222377":{"kind":"REWORK","_number":1,"created":"2022-05-05 16:13:26.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/96/840696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/840696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/840696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/840696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/840696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/840696/1"}}},"commit":{"parents":[{"commit":"3b1068eef8271653f8907bac62a587d8702e85e7","subject":"Merge \"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b1068eef8271653f8907bac62a587d8702e85e7"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-22 09:02:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-05 16:12:34.000000000","tz":120},"subject":"Reject invalid whitespace in HM url_path value","message":"Reject invalid whitespace in HM url_path value\n\nPlain whitespace characters in url_path must be rejected because\nthey may allow authorized users to inject arbitrary HAProxy\ndirectives in the configuration of amphorae.\n\nStory: 2008994\nTask: 42656\nChange-Id: I0f4c59a2928f2a813171109aaf73a1d7dff9eefe\n(cherry picked from commit 251fab17b51f99f1f0c1cce4985693993d8fb957)\n(cherry picked from commit 9afb3af6bb44cff6b0b0d92ed8b87c910cab4092)\n(cherry picked from commit 04207f60018b1b42dc81c5e4a8dcf0e98612bd21)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4503bd2767a33665c3474e4aa67ff45824222377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4503bd2767a33665c3474e4aa67ff45824222377"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
