)]}'
{"id":"openstack%2Foctavia~840698","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I0f4c59a2928f2a813171109aaf73a1d7dff9eefe","project":"openstack/octavia","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-05-07 10:41:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0f4c59a2928f2a813171109aaf73a1d7dff9eefe","subject":"Reject invalid whitespace in HM url_path value","status":"MERGED","created":"2022-05-05 16:14:00.000000000","updated":"2022-05-07 10:45:09.000000000","submitted":"2022-05-07 10:41:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840698","meta_rev_id":"0901374da2f5465139e47ea66768f6f945ac0722","_number":840698,"virtual_id_number":840698,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-07 10:41:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-05-07 08:03:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-05-05 16:45:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-05-07 08:03:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 16:45:23.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-05-05 18:09:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-07 08:03:55.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"9ae206086ff0d6fb2d80893b83c453c7742efec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-05 16:14:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95730f2c22a9197e8ebdc2d2c2320e869592133a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-05 16:45:23.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9a813699b2623b150e2f41cbbf53d1560cbe691","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 18:09:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06d49b2d9700443e8190e3dfe2ddd7f6 : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83839c495bad448987f11fe046b1020e : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a5eed41b681400bbaf65a92ef2babf6 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f81507d297440b0a0b6873e64b172e6 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32ef3b5c6ce24598bfbd313e49a207e6 : SUCCESS in 12m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36d61281315d409ea822a27bad494d7b : SUCCESS in 6m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ac041431121a46978bc7762b50b6b5d5 : SUCCESS in 5m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/44e47f54938b4495871028b1ef09919c : SUCCESS in 8m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/55d04f36e3a241f6ab5f05c4b89a4ea5 : SUCCESS in 6m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/122af56dcc744d16830e34563515d59e : SUCCESS in 1h 09m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac370e8bd18941788c2a2d11242a4f1b : SUCCESS in 1h 27m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5fb039769f4d47cf8b69062042243679 : SUCCESS in 38m 27s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4cbc2b7875224dafbd7b2c940e6ae5fd : SUCCESS in 36m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/365d5219c01c4bdda9ba916719122168 : SUCCESS in 37m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d0f80686a3904ffc9f62468ca17c1d15 : SUCCESS in 1h 22m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/68273de97f49495d94b49a5cfbd1da7d : RETRY_LIMIT in 6m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9c62af44253d4adb8e837abd243c4cf6 : SUCCESS in 1h 51m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79e485aa4fe95fa966970311d5ddb7ddc4e9f454","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-05-07 08:03:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"651cfe18fa5122713aab243e8e2ce61d36ab39c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 08:04:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af5b5e3b013e178b7905a47a3aa3f84e9e5b13b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 10:41:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ababc76bc722448c917b2e83977de45b : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ace37496cbfd4c828145b22c4dd636ab : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a975f02a9862498182d563ad35cf7bc5 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d699a7d1461e46a4b1a8bb4daa5d2d6e : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78eece756fba4f7895c6010b45305917 : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/295e873ce40248c7bcc7990bf7641454 : SUCCESS in 6m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7eca0a182e0844aaafc6f39ed37eb953 : SUCCESS in 1h 11m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef853452aa7145259704284c8a70e34c : SUCCESS in 1h 46m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/18cfc60e8aed49baa082823f4bd676e1 : SUCCESS in 37m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1b2ecf5d481741a1b746cb1b0a9c9d44 : SUCCESS in 32m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c483cae3c96c414d93617bf1a273e6f5 : SUCCESS in 39m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a15aed67cd0ea0edee8f26468887d19105d9007c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 10:41:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0901374da2f5465139e47ea66768f6f945ac0722","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 10:45:09.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efbfdeee9c37446487c0650c09ca07cd : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34cb4df87d824383bb0b59dc3385c3e4 : FAILURE in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5acee9e39f3fb1bad7e897f085c6fe5eb6826343","revisions":{"5acee9e39f3fb1bad7e897f085c6fe5eb6826343":{"kind":"REWORK","_number":1,"created":"2022-05-05 16:14:00.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/98/840698/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/840698/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/840698/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/840698/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/840698/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/840698/1"}}},"commit":{"parents":[{"commit":"82b2620b55f248fc1077123a68fd19b8b786e10c","subject":"Merge \"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/82b2620b55f248fc1077123a68fd19b8b786e10c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-22 09:02:24.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-05 16:13:49.000000000","tz":120},"subject":"Reject invalid whitespace in HM url_path value","message":"Reject invalid whitespace in HM url_path value\n\nPlain whitespace characters in url_path must be rejected because\nthey may allow authorized users to inject arbitrary HAProxy\ndirectives in the configuration of amphorae.\n\nStory: 2008994\nTask: 42656\nChange-Id: I0f4c59a2928f2a813171109aaf73a1d7dff9eefe\n(cherry picked from commit 251fab17b51f99f1f0c1cce4985693993d8fb957)\n(cherry picked from commit 9afb3af6bb44cff6b0b0d92ed8b87c910cab4092)\n(cherry picked from commit 04207f60018b1b42dc81c5e4a8dcf0e98612bd21)\n(cherry picked from commit 4503bd2767a33665c3474e4aa67ff45824222377)\n(cherry picked from commit 858bfd7129a67ce175591be15479dcf7512ec3e2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5acee9e39f3fb1bad7e897f085c6fe5eb6826343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5acee9e39f3fb1bad7e897f085c6fe5eb6826343"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
