)]}'
{"id":"openstack%2Foctavia~842052","triplet_id":"openstack%2Foctavia~master~Ib3d93b0897014d8def2599fabaf3a24c8f903849","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ib3d93b0897014d8def2599fabaf3a24c8f903849","subject":"WIP Remove device_owner param from VIP ports","status":"ABANDONED","created":"2022-05-17 08:23:52.000000000","updated":"2024-06-06 09:26:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ff0c6b89995241e92ef6574e5cfadd0603cd66aa","_number":842052,"virtual_id_number":842052,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-17 10:58:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-17 10:58:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba4cd1cc8547d57d705080525e579554a6c220df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-17 08:23:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb140ffdb4f9cc3b9f6f408e35edc8027a52c159","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 10:58:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66cbc6cdfe384af0b3cc6bebcc807d53 : SUCCESS in 12m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d38a0d98fae44e9b8169d1e0cf554e09 : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c1f0540daae4c7cbc63d9f415cd3a95 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dc4929af7fa43e0bdbd24e51a8c61e5 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea9133ff453a44978adedd683cf4fe92 : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c029ac4b04c4460897ef61debd19f98e : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b24dd45121d548d7bfbf2c39d35b53f9 : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d64b62471a184225b2dff39c4d18d355 : SUCCESS in 5m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6bed1408b35c437eac9fbe132d24c25e : SUCCESS in 5m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0a6e3de305d24b84879f9b2a5b5491cd : SUCCESS in 10m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/85a2a6fd749345688fbf163a33cacd3c : SUCCESS in 6m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f8d51d90b9314e22b706db3d21f9665c : SUCCESS in 8m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/789ceef137034a8bb2524952e098e34e : SUCCESS in 9m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/45f747d75e0f4af59dce4e4ebbb649c0 : SUCCESS in 8m 03s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/10dc30042a1b411487b3ca74b4dd437d : SUCCESS in 2h 06m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f169e514b234db3a2a4c7efaf8bb93d : SUCCESS in 2h 05m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/68729951233d4fe99611dc30241a4c02 : SUCCESS in 44m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/51f93d3957324b18a3eeec428f22a4a7 : SUCCESS in 1h 14m 22s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/daafa75a69004e2d80583bbd02da02ef : FAILURE in 17m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e52d15b3f162424fbbff1dd82ee9bc74 : SUCCESS in 43m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b2eee451a44d4bd49dea5247f6e325bc : SUCCESS in 2h 16m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/40acf3cde9774415b1fbbea9ff6aeadb : SUCCESS in 2h 21m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/22e8a25c9f0042d4a958b7ada59aba36 : SUCCESS in 2h 13m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b897c35bf5f443e29e74a26f31eb248c : SUCCESS in 2h 02m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8a457caa01fa451581c3e9a671334e27 : SUCCESS in 1h 54m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b755abcc3d4448c9e53b1fc2c483cb2 : FAILURE in 16m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff0c6b89995241e92ef6574e5cfadd0603cd66aa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-06-06 09:26:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa19b20bab5da9e8063af511370bf52ee24158c0","revisions":{"aa19b20bab5da9e8063af511370bf52ee24158c0":{"kind":"REWORK","_number":1,"created":"2022-05-17 08:23:52.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/52/842052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/52/842052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/842052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/842052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/842052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/52/842052/1"}}},"commit":{"parents":[{"commit":"a94b3101e8005ddd84a4333aa237dbbe3e0c2b43","subject":"Merge \"[doc] Fix client key name in command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a94b3101e8005ddd84a4333aa237dbbe3e0c2b43"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-05-13 17:53:46.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-05-17 08:19:49.000000000","tz":120},"subject":"WIP Remove device_owner param from VIP ports","message":"WIP Remove device_owner param from VIP ports\n\nML2/OVN does not support VIP ports that have a non-null device_owner\n[0], if a device_owner is set, OVN might reply to ARP requests to the\nVIP address, breaking the communication with the VRRP ports of the\namphorae.\n\n[0] https://github.com/openstack/neutron/blob/\\\n    c87ae2c0c68d4dad4c558a1b74d3a71bf820c324/neutron/plugins/ml2/drivers/\\\n    ovn/mech_driver/ovsdb/ovn_client.py#L301-L308\n\nStory 2010040\nTask 45384\n\nChange-Id: Ib3d93b0897014d8def2599fabaf3a24c8f903849\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa19b20bab5da9e8063af511370bf52ee24158c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa19b20bab5da9e8063af511370bf52ee24158c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
