)]}'
{"id":"openstack%2Foctavia~842564","triplet_id":"openstack%2Foctavia~master~Ie02b5164f3c1089f0d48101aab9e007bb4d132ea","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-09-02 16:30:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie02b5164f3c1089f0d48101aab9e007bb4d132ea","subject":"Update zuul queue configuration","status":"MERGED","created":"2022-05-19 16:18:50.000000000","updated":"2022-09-02 16:32:22.000000000","submitted":"2022-09-02 16:30:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842564","meta_rev_id":"bdcaf80002a3b5f4f7a32b3018ac8e3811a86186","_number":842564,"virtual_id_number":842564,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-02 16:30:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-05-19 16:24:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-09-02 14:06:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-07 14:04:49.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-09-02 14:06:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-05-19 16:24:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-19 19:01:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-23 07:54:49.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-09-02 14:06:00.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"8b135b74f3c965d92dfac224c5badf1f00b326b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-19 16:18:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e215b479d189e624384d338a936e7605b100cbc3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-19 16:24:26.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"5b3ced2f6629e15b93a14f7392ff3e3e9a50e5a3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-19 16:24:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a436a95aeb51114f6936de247eefb6a1ad95f37b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 19:01:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b75a748443a497db262c4d042bf3a0b : SUCCESS in 9m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6da49e29380049ef8907f490e6e3d65b : SUCCESS in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec605a3491d9449a8a341feacc7bb6c9 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2822e5e218ff4d7bb847394e8f534cc1 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ede44be68174e63af64d609f53ecaed : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/615e109065b645d8b69228a44db4b0f1 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/467db64fdf7b4aa0a97bb3040bca5044 : SUCCESS in 15m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2983a83417844b3a88675fd1c26db9ed : SUCCESS in 6m 05s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4ba4442d510b4230bb9d550cf17db3f3 : SUCCESS in 10m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c3f5bae63188412285b0b25d02af2376 : SUCCESS in 6m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ce5a198cc9dc47638e459ea5a9480f0b : SUCCESS in 8m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/13752901d4ff4fa4bf9be9e656bfeaf5 : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4358f19e9f19434b803f213bd5e2364f : SUCCESS in 8m 04s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7a025560983b493eafc39037420b16c2 : SUCCESS in 2h 21m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/adfb7f9dffc64f33a7d5c32624d3d8c9 : SUCCESS in 1h 55m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/569b8592ee0c43849bdee192abed47a1 : SUCCESS in 36m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4263219620dd45ccb5d8118a803b58a0 : SUCCESS in 52m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/af2034ac15ff489b8665b6277ccce6c0 : FAILURE in 18m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/34ac62f09121491b834c355e73218dd2 : SUCCESS in 37m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9312784cf05f4a1ca5632a73227a25ad : SUCCESS in 1h 35m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/af03d647d0814c829a6c88c9d2d2695a : SUCCESS in 2h 32m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2f0e3bc8f1cb4a82be8776013458a3d9 : SUCCESS in 1h 28m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8a1d1e38982749d09e404ec93f4c340b : SUCCESS in 1h 59m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/69c69f05d77e44c68657ee0d62b1bb4f : SUCCESS in 1h 44m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/95316d2c8c0b41f7bc5bfd9ae504e3c6 : FAILURE in 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b07973736cd20d34fce30f97894c098582226d0","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-23 07:54:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2c5b2bd61b6589c46b8c13e71996d9e638125942","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-07 14:04:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2c27ded56b53987cffeff57f294a67b518a0b968","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-02 14:06:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce02e03396ac69bb4b1372fdf3beb4280c85b93d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 14:08:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c46ca92ce17a25cab5cece6ce02b3b4a71a64752","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 16:30:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fa1369c2c2346b085db5c0af99bf600\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73378affb6d946da98a81b33af8bffd6 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06e2b99609a24db3a1cb107db7417763 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d0fe99ae1bf402b91563f4118163ff8 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79d23501276142dd85b87a5b28453ce7 : SUCCESS in 11m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f30984c3305149b28db20317e22ce3f8 : SUCCESS in 5m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fa2678ce2d4846b98287905227f2cc64 : SUCCESS in 7m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7ff423bc23de4176abeaa58cf1147e19 : SUCCESS in 2h 07m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1fbff1a17224477c8b8acd004a1afe0d : SUCCESS in 2h 18m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7d43f539936f4dbba0bdbb934558b2f6 : SUCCESS in 45m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/81d9700b8bee48b7a2a7810c362d0f6d : SUCCESS in 1h 10m 16s","accounts_in_message":[],"_revision_number":1},{"id":"abf02cac4f633c33e73a4a570901647887d7b97a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 16:30:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bdcaf80002a3b5f4f7a32b3018ac8e3811a86186","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 16:32:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/532316f1b53246df9a1dcf62865eaa2c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/871b0f4506c443c0a064f4eaf120720b : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81105bbaed9e83e98c46e70c02b1bd43c4a9d875","revisions":{"81105bbaed9e83e98c46e70c02b1bd43c4a9d875":{"kind":"REWORK","_number":1,"created":"2022-05-19 16:18:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/64/842564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/842564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/842564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/842564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/842564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/842564/1"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-19 16:18:02.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-19 16:18:49.000000000","tz":0},"subject":"Update zuul queue configuration","message":"Update zuul queue configuration\n\nZuul is changing how queues are declared[1]. This patch updates Octavia for the new way to define the queue.\n\n[1] https://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nChange-Id: Ie02b5164f3c1089f0d48101aab9e007bb4d132ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81105bbaed9e83e98c46e70c02b1bd43c4a9d875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81105bbaed9e83e98c46e70c02b1bd43c4a9d875"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[]}
