)]}'
{"id":"openstack%2Foctavia~843187","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~Id0ba1d06600e025e3825d6c4a2a881b4f7f8d153","project":"openstack/octavia","branch":"stable/victoria","topic":"bug/1975036","hashtags":[],"change_id":"Id0ba1d06600e025e3825d6c4a2a881b4f7f8d153","subject":"DNM: Testing https://review.opendev.org/c/openstack/tempest/+/843182","status":"ABANDONED","created":"2022-05-24 19:43:37.000000000","updated":"2022-05-31 16:29:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1665e4d5fc6e4b94b0ef2882eaf604300d759716","_number":843187,"virtual_id_number":843187,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-24 21:24:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-24 21:24:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e67bff7b966cf6dfb53f5d3faca2668dfcc6c073","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-24 19:43:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b390a54d5bafa9493a794da9d2adc3283c90bf5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-24 19:45:34.000000000","message":"Topic drop-py3-6-7 removed","accounts_in_message":[],"_revision_number":1},{"id":"9fef9c2a452b09aedccba63446d33742a0ba69b6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-24 19:45:36.000000000","message":"Topic set to bug/1975036","accounts_in_message":[],"_revision_number":1},{"id":"aa3a8d16431db42c0fac493a4dccb0119d2d8b48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 21:24:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bba5d24ce9584be4ba4eb01484962deb : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/641218620d60464699bbe0fa5905a496 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a1f67c4fa754a019a0862114c07f02b : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/694cdb6cd64d4717bdfbc229a0f5667c : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31534e6f1bf4b8d9f8de8b3b6a2bc04 : SUCCESS in 11m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8cb10177ac574c8b91db67e75e198832 : SUCCESS in 7m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4635632bcdbb43e9930c21f5768108eb : SUCCESS in 8m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b06b0fd30bbc441ea9f4d9fce16daaf9 : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/81e255d26bf84519ab9dafcafe66a496 : SUCCESS in 6m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e5d74203d8049b99b3b0e2167fddd2a : SUCCESS in 1h 17m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02e330577ce5407a9da212a334891e93 : SUCCESS in 1h 25m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/972f3be5fee742348d0f8ab5e1972bda : SUCCESS in 35m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ff0045a4e145444d9ca83644194101e6 : SUCCESS in 32m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/990ff117a7214c64a666d728a9e17285 : SUCCESS in 34m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3107cbfd096849d792d6f2aeaf3f22ad : SUCCESS in 1h 31m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/571ad165443e4b508b0335104e02bf06 : FAILURE in 29m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/863da1969bd84ee2a67cda23a2c98427 : SUCCESS in 1h 29m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b721ffb9586d45c3ae2a8f757abddb86 : FAILURE in 1h 32m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fc6d95e759354c8b89f73c18fc49f733 : FAILURE in 1h 35m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1a1bfcd039ce416c8f036ef21e2be1d6 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1665e4d5fc6e4b94b0ef2882eaf604300d759716","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-31 16:29:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e62e52bacefe465107283fb9ab4f26f4fb474c8","revisions":{"2e62e52bacefe465107283fb9ab4f26f4fb474c8":{"kind":"REWORK","_number":1,"created":"2022-05-24 19:43:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/87/843187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/843187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/843187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/843187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/843187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/843187/1"}}},"commit":{"parents":[{"commit":"0f595fbed0cc4d1afbb4d44d69ae9896bf7771bd","subject":"Merge \"Add keepalive for redis-based taskflow boards\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f595fbed0cc4d1afbb4d44d69ae9896bf7771bd"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-24 19:40:16.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-24 19:40:16.000000000","tz":-300},"subject":"DNM: Testing https://review.opendev.org/c/openstack/tempest/+/843182","message":"DNM: Testing https://review.opendev.org/c/openstack/tempest/+/843182\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/843182\nChange-Id: Id0ba1d06600e025e3825d6c4a2a881b4f7f8d153\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e62e52bacefe465107283fb9ab4f26f4fb474c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e62e52bacefe465107283fb9ab4f26f4fb474c8"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
