)]}'
{"id":"openstack%2Foctavia~843296","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I1c7289ce9e765c71f5353ffd1d367d33cc86a144","project":"openstack/octavia","branch":"stable/victoria","topic":"victoria-pin-tempest","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-05-25 21:40:02.000000000","reason":"Ghanshyam Mann replied on the change"}},"hashtags":[],"change_id":"I1c7289ce9e765c71f5353ffd1d367d33cc86a144","subject":"DNM: testing Tempest pin in stable/victoria","status":"ABANDONED","created":"2022-05-25 15:35:27.000000000","updated":"2022-05-31 21:31:13.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"31266311bb1866a3a4273c0c2301620b302cee20","_number":843296,"virtual_id_number":843296,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-25 23:56:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-25 16:02:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d3aae7048e9078d0ae6c5aeb7e2761ad55239215","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-25 15:35:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"992b0b29a0ae1152af3164ccc7253ecbead514c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 16:02:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5aebc42f9674084acda06626ee9ed1e : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f515030ead75496ebabc68a5e81dfa1e : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/185297f6123d4f84a7def09e0d6abda5 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/309e56a7190b478dae71c0e418bb02b7 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ff0563ba2a540048f7d87334f1079aa : SUCCESS in 12m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9b12377b53e0457bb7e81287bce3bea6 : SUCCESS in 6m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c788587fdf534c8cae2ddbc5b7b79c7e : SUCCESS in 9m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/faff8614903b40fc82ac900de00bd9cb : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9668f765a6f644059bda6b52d5a4346f : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/549cef7820904c0ca1e5f4fa9639e8b3 : FAILURE in 22m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba9d89fd0a1a44d99125b00f763577f4 : FAILURE in 12m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0441070fecc34d6aa9933f753a562670 : FAILURE in 12m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/db37ad7e61e14d94aaa9ac1e367c49e5 : FAILURE in 12m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/05c75c85aced48a0b2af85f660a26fb9 : FAILURE in 12m 17s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2ae8f33f64b6459a95b562869c394e55 : FAILURE in 13m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5baf810eb6c3485eba434e8874ff1cde : FAILURE in 16m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8aad9e0283d646fd961fdb52850f1054 : FAILURE in 17m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/125c77f1cacc457ba81ae0985faab913 : FAILURE in 13m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/87df1cfd743a44ccb36c1f2d273ea3f4 : FAILURE in 13m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6c1aa032c925407d86dbf56906265fde : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3f10f376b9f3eaa3b939e24846fa7b25d04a0b6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-25 21:40:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"12cb8123d8d4647355b2dca0d7f64bb9e8d97535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 23:56:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/972132facde4431499b5ce4d87ab3910 : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ba3122a9bac46dbbce64685f81ae705 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/773eadacd268497fa77f4ac2713e8fa3 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea557541fa214184b5762216345f2e90 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d610a181a5e4403c85c41b9acdbdfbc2 : SUCCESS in 13m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c7360f0b3ce04a32aedf9d223d781606 : SUCCESS in 6m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c5a6e02b3a7e45dcb6ad10049c52a999 : SUCCESS in 8m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bbdefb89f36b43c19ae8669992e4706a : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/55ac9696272141c4b858717e93763f3e : SUCCESS in 6m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7abf54fd3c12469c8913f4eafc173ca5 : SUCCESS in 1h 30m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/147288f07e384bdaa151bd300116c276 : SUCCESS in 2h 11m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/df1b573391864850a6986fe85da04755 : SUCCESS in 43m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3fe76d6d5642465faa06ad365630421e : SUCCESS in 32m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45c4ade027f24e7cbd6e39d4aeab5c8f : SUCCESS in 33m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/05f313664d274de68c013c887e61990b : SUCCESS in 1h 36m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/78b644e19d1f440dbe525f14e367476d : FAILURE in 30m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/45d0e9c1c734486988f9fc0a8334746b : SUCCESS in 1h 26m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/03b066a424c941e684302cd109cbe470 : FAILURE in 1h 33m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/952e919ea41e4692be9319d10ddd0534 : FAILURE in 1h 50m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/66361f7032af421294bf954740b2112b : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31266311bb1866a3a4273c0c2301620b302cee20","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-31 21:31:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d8f892a89deea93cb80018e6de116b67a2ab0d0","revisions":{"2d8f892a89deea93cb80018e6de116b67a2ab0d0":{"kind":"REWORK","_number":1,"created":"2022-05-25 15:35:27.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/96/843296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/843296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/843296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/843296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/843296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/843296/1"}}},"commit":{"parents":[{"commit":"0f595fbed0cc4d1afbb4d44d69ae9896bf7771bd","subject":"Merge \"Add keepalive for redis-based taskflow boards\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f595fbed0cc4d1afbb4d44d69ae9896bf7771bd"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-25 15:34:46.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-25 15:34:46.000000000","tz":-300},"subject":"DNM: testing Tempest pin in stable/victoria","message":"DNM: testing Tempest pin in stable/victoria\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/843295\nChange-Id: I1c7289ce9e765c71f5353ffd1d367d33cc86a144\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d8f892a89deea93cb80018e6de116b67a2ab0d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d8f892a89deea93cb80018e6de116b67a2ab0d0"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
