)]}'
{"id":"openstack%2Foctavia~843306","triplet_id":"openstack%2Foctavia~master~Ib60ad9a52735c0ceb3cddeebab856de81cc8157b","project":"openstack/octavia","branch":"master","topic":"victoria-pin-tempest","hashtags":[],"change_id":"Ib60ad9a52735c0ceb3cddeebab856de81cc8157b","subject":"DNM: testing Tempest pin in stable/victoria","status":"ABANDONED","created":"2022-05-25 15:54:15.000000000","updated":"2022-05-31 21:31:08.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"666a38834f9ae365786a53a3d7f139ddd103f32a","_number":843306,"virtual_id_number":843306,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-25 18:35:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-25 18:35:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"792959d25084370cff686e182333ec3e571ab9ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-25 15:54:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba95bc8f049ee63faaac2074c1ef066a9f1b20b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 18:35:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f756ae94c7e44259a51b5d34195d7af8 : SUCCESS in 13m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9278331c6ca24748b8d76e3d8ff86ae1 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc79db1cf5ec4bcfa322350779a50789 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/439ea27850674021ab5ee5aa2101dcb9 : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77f9a2dff7e04fdeaa3838c7e35430ff : SUCCESS in 7m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/978fecc727d945b787d6606c047c74da : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9f1038f5c6b4bccbd5c37e27bd894d1 : SUCCESS in 13m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/21ff6050d6a84e7db93163aa2f4fbfc8 : SUCCESS in 8m 02s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/52d773ae75fa495c8dc9014c404ff84d : SUCCESS in 10m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a89604d1dfad4f7c9f2202dc2c293a70 : SUCCESS in 6m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4a3ce9da54f347099379a6fe55ebf21d : SUCCESS in 9m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/59567cebee794de2addbca836f76a755 : SUCCESS in 9m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4166168f5e4141b68ef1a8e5c0d22fa1 : SUCCESS in 9m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/96e0b1df18ec4c41b24a23a2b13024cd : SUCCESS in 2h 33m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/38001df3af4a4eeaae29968bbfd47991 : SUCCESS in 1h 37m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/20b9c25cda2e40e8982f42070512b9b3 : SUCCESS in 34m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ce728117d10747ca87a9ad3fac8dafb1 : SUCCESS in 53m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/57cd39ad13164b29a4fd354dbab2ce8f : FAILURE in 18m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1cad16fd2c814909ba88cca4a39bb991 : SUCCESS in 37m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/10335679d24d4528a95c4fa654550222 : SUCCESS in 1h 49m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c8c8bae251cf4c6e8d9d4f923a53b90a : SUCCESS in 1h 44m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6696bc77784b4cecb3072e978d97994f : SUCCESS in 1h 51m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/69d4b9249f1347bab21415256c23d28c : SUCCESS in 1h 34m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8a806c11790f4fbc8d6c350fd60e24a2 : SUCCESS in 1h 40m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/c573317d1ccf433f999573ff2f7e3864 : FAILURE in 15m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26b1843007355897c522b4cd46ba946cd3cd8023","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-25 21:40:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db9450b22fdb007fa79b81bec61473bcf58a3d15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 23:58:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af3a5d8fadb94b36b1f2660748e4414c : SUCCESS in 12m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1833cde6f3044b17b6243552f71c04ee : SUCCESS in 12m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13bb014efdc645108e261f55b730fdb5 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1542cc7b79404cfda22978286f68e45f : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7065d18e5d084023887b318e6dfd11e7 : SUCCESS in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad1af615e86d46889517d93130577c44 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f260aef2b9241c69aae338c4d6325ee : SUCCESS in 14m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bb5834d8ae9c4490a319149f8c15221f : SUCCESS in 7m 39s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/46804f8c6030423287ecd41db7d13f84 : SUCCESS in 9m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5462b28f3f9d473cb7ba8544c10dec9a : SUCCESS in 6m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/786ac141703f4ce4b318354fdaeeeb1a : SUCCESS in 9m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/51465a549c3d44a596045e5c282c3a4f : SUCCESS in 9m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5cf0f00055e946f8824522d0d4165dc7 : SUCCESS in 9m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7baf7230bfad43ff88e3c3a3555fa67a : SUCCESS in 2h 10m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/224b7a67d4184eb3aefd72e4027ce874 : SUCCESS in 1h 25m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e6ec96315bed468384195c7783a16c80 : SUCCESS in 35m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/78810f2f76a04b0aa1afd7ffdcec63c5 : SUCCESS in 55m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/dc835a89e2df48418f071d1ef7c304aa : FAILURE in 18m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/991b74dda17b4188ae85b9614cd3bde6 : SUCCESS in 39m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1955fab38d244eab8afa15e8b634a1f2 : SUCCESS in 1h 32m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0052679b78724b24b55a137ba3047141 : SUCCESS in 1h 43m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cc33bdf8368d4d9aac9d853146617dc2 : SUCCESS in 1h 37m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d5edda7b2b85438183e3300e6eb7bf9f : SUCCESS in 1h 32m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ceb3d37f44264e1fa7edbdbcdf548f81 : SUCCESS in 1h 21m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d022d29aaf4461da0b7463d34d8245e : FAILURE in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"666a38834f9ae365786a53a3d7f139ddd103f32a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-31 21:31:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8e6a17003d1faea75f47c74dc3687ea06b1a23f","revisions":{"c8e6a17003d1faea75f47c74dc3687ea06b1a23f":{"kind":"REWORK","_number":1,"created":"2022-05-25 15:54:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/06/843306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/843306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/843306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/843306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/843306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/843306/1"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-25 15:52:43.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-25 15:52:43.000000000","tz":-300},"subject":"DNM: testing Tempest pin in stable/victoria","message":"DNM: testing Tempest pin in stable/victoria\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/843295\nChange-Id: Ib60ad9a52735c0ceb3cddeebab856de81cc8157b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8e6a17003d1faea75f47c74dc3687ea06b1a23f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8e6a17003d1faea75f47c74dc3687ea06b1a23f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
