)]}'
{"id":"openstack%2Foctavia~843838","triplet_id":"openstack%2Foctavia~master~I8072e000fac02166d989e7fdb358c4dc34fa1bfa","project":"openstack/octavia","branch":"master","topic":"centos-9-stream","attention_set":{},"removed_from_attention_set":{"6469":{"account":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"last_update":"2022-06-21 14:19:34.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-02 20:03:33.000000000","reason":"Michael Johnson replied on the change"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-06-21 14:19:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8072e000fac02166d989e7fdb358c4dc34fa1bfa","subject":"Update default image versions for CentOS and RHEL","status":"MERGED","created":"2022-05-30 10:39:28.000000000","updated":"2022-06-21 14:22:33.000000000","submitted":"2022-06-21 14:19:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"843838-centos-9-stream","meta_rev_id":"a9fc1bcbb36677e853e9b09e2732d5266b4b8038","_number":843838,"virtual_id_number":843838,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-21 14:19:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-06-02 20:03:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-06-21 11:59:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-06-21 11:59:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"CC":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-30 12:59:46.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"CC"},{"updated":"2022-05-30 17:23:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-31 18:32:26.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-06-21 11:59:53.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"f6319a68eb9b85a695773872c09017ef32a3828b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-30 10:39:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3eb430208ce9bdf628037629d6f03f157e62c71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-30 11:43:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"64b37c56f915103779c2c1af99c771c9c2f8a1f5","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-30 12:58:26.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"18ce3a05cef4d62f737f5f04103d06ff3b5b54e3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2022-05-30 12:59:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91307db7b5193ecec946fd411da7aca934619364","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-30 13:34:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"629fb979c6ad03228b1b3fa2ae758f7efa57f5ba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-30 13:41:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"953a31a5f0d2058fea007faae128b572ea578b61","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-30 14:04:23.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"d76c55e8beb40ac0a5658ef7063f095059875aef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-30 15:17:03.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"1ac85d830542dff343718499cde6fb3268f32d8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 17:23:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/396f8b7e53eb4da8814c6984e727af71 : SUCCESS in 11m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ca128de4d3b45f5aec76ab9686afc0b : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de09fa85efb74654b22b878d7bc96d90 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c78e19c236448bca1a893954788df67 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69d96717269441cd9323bacdeb086d5d : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4c09a68faa249119d06b820b5c20116 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2354937ac69c4d3eb70d7155ad3d8251 : SUCCESS in 12m 12s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1b2526fc2a0c4ba9a99e3e62e49d9f54 : SUCCESS in 6m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e0ff2be7d268491d9add1f74cf9c95b3 : SUCCESS in 8m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5cde8fcfd201468db27fa9d09ce1b4cb : SUCCESS in 6m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/18913adb957d491bb8bfd2a36e45bff9 : SUCCESS in 7m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d985061252f24789930c62f73bdd51b5 : SUCCESS in 8m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a3eb23ab81774cc9bb0ac02d8a631b2b : SUCCESS in 8m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2f55af753b0240d681864fc00c50e2bb : SUCCESS in 2h 03m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d4cf39d5436e45fa8f44a843c1c37fa8 : SUCCESS in 1h 29m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2330114fdf034273976c4c2e962d75cb : SUCCESS in 39m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/77d5ae486d244a08bfaf516557bcfe85 : SUCCESS in 51m 02s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a15b6ca91c654b8ca246b915bcf4b003 : FAILURE in 12m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/718a2758b068422da21e094dbd3585d1 : SUCCESS in 33m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2358d7bd1d5847ec81c2955ae02d10f7 : SUCCESS in 1h 38m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/52e52181055d4f149a8926b052ebe8ea : SUCCESS in 1h 57m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/aa29113e20f14a1ba0e2ae04775621c2 : SUCCESS in 1h 38m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/19c8d4d98a3946b6adf60e5fd93a3c1d : SUCCESS in 1h 31m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/62ece32a41a14517834f3edb643cfe87 : SUCCESS in 1h 23m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/4fbfdf14d3004779a1dc03f38ee3add1 : FAILURE in 9m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6479908da1982f80fee436aec3b1df210e958755","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-31 18:32:26.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"792e94bff62d0a41f9a110425d19f99f3674929e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-31 18:34:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"69e9b14f5154339f32cb978addb2420eb61c4eef","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-01 07:09:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"44c983d7e2feacf79873bfaaefe623621c85c088","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-01 07:40:31.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"e5069b42b1e6e3840bd1c906d050c5874fdb565a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-01 07:48:02.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"74dfd0f0422f8e074d0be8803024ea98dad00490","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2022-06-01 08:15:25.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3d1a06508b3f4f19bff8564a3a9cd6e6a137f998","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-01 08:19:31.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f6380fc1b933420b4deddcb3d1dbc815290b558d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-01 08:20:26.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"881fe3bc42007acee24e1a19955afd85bdf236ed","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-01 08:26:54.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"231c5fa1296afae9a9fc72d763d8a73783232c17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-01 08:33:16.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"6e4e8f10497bff5fb0b314a1514f588ee2376647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 10:43:20.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1facded3c97043439e73c13e158b1780 : SUCCESS in 10m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8570605ab7f64a01bb2b9d16f3a97be2 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b414848e05184f8c9a3ec5ed55771edd : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c839df39c2045bd816b45a88d470877 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/441718bec0df4460ac8127a7feacb7bc : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e89d2d330ded43d192546bd4bdcddeae : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e49f4fd77dc7455589061b247d4e218e : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc7209ecd75948c18227bd9d8d4760de : SUCCESS in 7m 20s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5900607267f242468dca77deabda3ed2 : SUCCESS in 5m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1d8eedb607a14640af471743c8ec32d3 : SUCCESS in 7m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e43fc2e7b6a943f696e0e068e2e4b0be : SUCCESS in 7m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/97f0b00cfba44c6ca456277ca761880a : SUCCESS in 10m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5e4538ff39f9460a9b21815f8859ff5e : SUCCESS in 7m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ee72d258a7e44089a4ebbc5dcaa1566b : SUCCESS in 8m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f128c47f3eb84475a681ebf5708c3fae : SUCCESS in 2h 06m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac25ed08bbd44fa1a5d8ae5da2466af7 : SUCCESS in 1h 27m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f2c2a07b543b49d9bdb7743700ad02e5 : SUCCESS in 35m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/abb6ed2bcd6f49d8a58c3ab90d7fb447 : SUCCESS in 1h 15m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b891b72127084ff3b25dfb76ab50e56e : FAILURE in 12m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe0f1da1b91c4dd788e8bde8a1893ece : SUCCESS in 43m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/008b35e22c994445a2013d2737d46865 : SUCCESS in 1h 58m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ce80bdb301274fcf81ba5f1787a5aa22 : SUCCESS in 2h 05m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/730c4f6ff315465e9d2f77f9402c584f : SUCCESS in 2h 05m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3aa1174681ff4ce1b646ebc5c5451010 : SUCCESS in 1h 32m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b3d6c2261da94252b67f500f228379eb : SUCCESS in 1h 35m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a94bea58d9e4fb485ed3bf4797fd6f4 : FAILURE in 10m 47s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"66d0e31fa034f1783e60c4a57ec2e6a07511365e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 20:03:33.000000000","message":"Patch Set 10: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"6e07be4edbf693702b481d4f57058e3d29b9584a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-21 11:59:53.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"8fd8af66ccab3bad28347584f6db7b8797ac49c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 12:00:47.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"77e352942054d72145c72cc03ef734a5eaf24e5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 14:19:34.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2592eb93d9748dda7130f461fcd3852 : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c023d93bd06454ba9293d0377ae5d79 : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/781581d2c1264f109a85206284ec9653 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87fb3e1217494a42a1fc1799543d3e36 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b83f607332d413698751ba1b588484f : SUCCESS in 13m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e2df8a537cc4714b77a719708519006 : SUCCESS in 4m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/91e8b93a8b794aacab0378ce54e546d5 : SUCCESS in 6m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4ea824a8858641daad4c9b9685b4d946 : SUCCESS in 9m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/947ef2a0459a4117887b332d18a464c7 : SUCCESS in 2h 12m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07ad26c2656b49a8a5b7ce993f6644f8 : SUCCESS in 1h 40m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1b56c5fdf54b43adb8f67135ccf11733 : SUCCESS in 34m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ba99119d3e7341eda5b470849e8f22eb : SUCCESS in 52m 02s","accounts_in_message":[],"_revision_number":10},{"id":"7239b79ceafa70b7aaf7ff715e3f1a4a1514854b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 14:19:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"a9fc1bcbb36677e853e9b09e2732d5266b4b8038","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 14:22:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af8fbf7710544e6db6ec908889381ffc : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/828a6afd318b40fea7dbb419f338d537 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"f444fa7c4d4e946d41f6a03b9bc02eb05584ef79","revisions":{"a695fd1687291b8005f59dcd4b8bb54753455e84":{"kind":"REWORK","_number":1,"created":"2022-05-30 10:39:28.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/38/843838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/843838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/843838/1"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-30 10:38:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-30 10:38:38.000000000","tz":120},"subject":"WIP use jammy release by default when building Ubuntu amp images","message":"WIP use jammy release by default when building Ubuntu amp images\n\nChange-Id: I8072e000fac02166d989e7fdb358c4dc34fa1bfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a695fd1687291b8005f59dcd4b8bb54753455e84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a695fd1687291b8005f59dcd4b8bb54753455e84"}]},"branch":"refs/heads/master"},"f9e2716907e0b77e2db02035ad1117eaed599d96":{"kind":"REWORK","_number":2,"created":"2022-05-30 11:43:03.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/38/843838/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/843838/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/843838/2"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-30 10:38:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-30 11:40:52.000000000","tz":120},"subject":"Update default Ubuntu release for amphora images","message":"Update default Ubuntu release for amphora images\n\nUse the latest Ubuntu release jammy as new default for Ubuntu\nimages.\n\nChange-Id: I8072e000fac02166d989e7fdb358c4dc34fa1bfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9e2716907e0b77e2db02035ad1117eaed599d96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9e2716907e0b77e2db02035ad1117eaed599d96"}]},"branch":"refs/heads/master"},"73f26cf28d0758ac2c2ca735d0a83c3d405e8624":{"kind":"REWORK","_number":3,"created":"2022-05-30 13:34:09.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/38/843838/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/843838/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/843838/3"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-30 10:38:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-30 13:32:01.000000000","tz":120},"subject":"Use CentOS 9 Stream as new default","message":"Use CentOS 9 Stream as new default\n\nAccording to [1] CentOS 9 Stream is an official\nruntime environment for the Zed cycle.\n\nChange-Id: I8072e000fac02166d989e7fdb358c4dc34fa1bfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73f26cf28d0758ac2c2ca735d0a83c3d405e8624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73f26cf28d0758ac2c2ca735d0a83c3d405e8624"}]},"branch":"refs/heads/master"},"69aa59dd6b7c03d4e48c6eb7d96800dd8443f3c0":{"kind":"REWORK","_number":4,"created":"2022-05-30 13:41:42.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/38/843838/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/843838/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/843838/4"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-30 10:38:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-30 13:40:01.000000000","tz":120},"subject":"Use CentOS 9 Stream as new default for diskimage-create","message":"Use CentOS 9 Stream as new default for diskimage-create\n\nAccording to [1] CentOS 9 Stream is an official\nruntime environment for the Zed cycle.\n\nChange-Id: I8072e000fac02166d989e7fdb358c4dc34fa1bfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69aa59dd6b7c03d4e48c6eb7d96800dd8443f3c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69aa59dd6b7c03d4e48c6eb7d96800dd8443f3c0"}]},"branch":"refs/heads/master"},"6b18c22655f561f04fc6a14dfc485cef5594e2bc":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-05-30 15:17:03.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/38/843838/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/843838/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/843838/5"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-30 10:38:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-30 15:16:20.000000000","tz":120},"subject":"Use CentOS 9 Stream as new default for diskimage-create","message":"Use CentOS 9 Stream as new default for diskimage-create\n\nChange-Id: I8072e000fac02166d989e7fdb358c4dc34fa1bfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6b18c22655f561f04fc6a14dfc485cef5594e2bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6b18c22655f561f04fc6a14dfc485cef5594e2bc"}]},"branch":"refs/heads/master"},"6e76cecc1a8b04ee1007b1a26032a9443e9f8082":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-06-01 07:40:31.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/38/843838/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/843838/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/843838/6"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-30 10:38:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-01 07:39:21.000000000","tz":120},"subject":"Use CentOS 9 Stream as new default","message":"Use CentOS 9 Stream as new default\n\nUpdate diskimage-create.sh\u0027s default for CentOS images from\nversion 8 to 9.\n\nChange-Id: I8072e000fac02166d989e7fdb358c4dc34fa1bfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e76cecc1a8b04ee1007b1a26032a9443e9f8082"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e76cecc1a8b04ee1007b1a26032a9443e9f8082"}]},"branch":"refs/heads/master"},"9b187042f48ecb385aa4c8e697c5760497798179":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2022-06-01 07:48:02.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/38/843838/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/843838/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/843838/7"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-30 10:38:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-01 07:48:02.000000000","tz":0},"subject":"Use version 9 as new default for CentOS images","message":"Use version 9 as new default for CentOS images\n\nUpdate diskimage-create.sh\u0027s default for CentOS images from\nversion 8 to 9.\n\nChange-Id: I8072e000fac02166d989e7fdb358c4dc34fa1bfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9b187042f48ecb385aa4c8e697c5760497798179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9b187042f48ecb385aa4c8e697c5760497798179"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b652e8eb0814fbe2fa027e6cf2f2ded0d862a8ea":{"kind":"REWORK","_number":8,"created":"2022-06-01 08:19:31.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/38/843838/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/843838/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/843838/8"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-30 10:38:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-01 08:18:59.000000000","tz":120},"subject":"Update default image versions for CentOS and RHEL","message":"Update default image versions for CentOS and RHEL\n\nUpdate new default version for CentOS images to 9-stream and for RHEL\nto 9.\n\nChange-Id: I8072e000fac02166d989e7fdb358c4dc34fa1bfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b652e8eb0814fbe2fa027e6cf2f2ded0d862a8ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b652e8eb0814fbe2fa027e6cf2f2ded0d862a8ea"}]},"branch":"refs/heads/master"},"68e007e2d0cddc136fd1b8008d91da75be70275b":{"kind":"REWORK","_number":9,"created":"2022-06-01 08:20:26.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/38/843838/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/843838/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/843838/9"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-30 10:38:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-01 08:20:19.000000000","tz":120},"subject":"Update default image versions for CentOS and RHEL","message":"Update default image versions for CentOS and RHEL\n\nUpdate new default version for CentOS images to 9-stream and for RHEL\nto 9.\n\nChange-Id: I8072e000fac02166d989e7fdb358c4dc34fa1bfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/68e007e2d0cddc136fd1b8008d91da75be70275b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/68e007e2d0cddc136fd1b8008d91da75be70275b"}]},"branch":"refs/heads/master"},"f444fa7c4d4e946d41f6a03b9bc02eb05584ef79":{"kind":"REWORK","_number":10,"created":"2022-06-01 08:33:16.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/38/843838/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/843838/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/843838/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/843838/10"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-30 10:38:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-01 08:33:07.000000000","tz":120},"subject":"Update default image versions for CentOS and RHEL","message":"Update default image versions for CentOS and RHEL\n\nUpdate new default version for CentOS images to 9-stream and for RHEL\nto 9.\n\nChange-Id: I8072e000fac02166d989e7fdb358c4dc34fa1bfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f444fa7c4d4e946d41f6a03b9bc02eb05584ef79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f444fa7c4d4e946d41f6a03b9bc02eb05584ef79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
