)]}'
{"id":"openstack%2Foctavia~845042","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I2261342377e7acb1d2bc4fc742a351dfc66e4305","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-16 10:45:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2261342377e7acb1d2bc4fc742a351dfc66e4305","subject":"Fix new pylint issues","status":"MERGED","created":"2022-06-08 14:48:07.000000000","updated":"2022-06-16 10:47:28.000000000","submitted":"2022-06-16 10:45:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845042","cherry_pick_of_change":845146,"cherry_pick_of_patch_set":1,"meta_rev_id":"7e8870949dfbd148e20015281c31fbceb2ae2f30","_number":845042,"virtual_id_number":845042,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-16 10:45:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2022-06-16 05:57:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-08 15:36:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2022-06-16 08:30:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-08 14:53:51.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-06-08 18:26:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-16 05:57:56.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"6553ba957acd07d1ce1dc7f445e05116e44fa5e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-08 14:48:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"66cb0121d02e6982098f9cf27d8662ed98037304","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-08 14:53:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f6add86d97b137e0a5005cc6d84122a05a6f34e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-08 15:27:19.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9b070e45c5d1719ed0c9750e69d8a246a95b14a4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-08 15:36:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"64200989d9939706791f4194948141447e149512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 18:26:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25026e9805314f96b6089ce6ec7fc140 : SUCCESS in 10m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b23774ad3e1c40aca3ce85bfb02fa032 : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09b38b9e269c4ad08a911586c972efc2 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7403eabddc04bb99bc6d723d9accba2 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5173c5875494ddca637f66847969764 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b9a560b81e144d79500838ae49b9354 : SUCCESS in 12m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6a2caa9289cf4ec19657c151c5cfbc90 : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f73fa2d5a2eb4801888b54f44d8ce55b : SUCCESS in 9m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/414b6c2488fd430eb46d8939597f648e : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c90d5259c5c240f4848e363d42a4f39b : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d9ea3a898f3488facb782720cd567fc : SUCCESS in 2h 57m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5eb3691de87445e90c5d004900214cd : SUCCESS in 1h 21m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c6076331c9dc447387a9a445bbb51b88 : SUCCESS in 34m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b59e805f75d641ca944ffc7facd64548 : SUCCESS in 46m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1ffec4b43b304921be6dac6a6c1f2e6e : FAILURE in 34m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74afb644eebd4239aef354b7c184e596 : SUCCESS in 36m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6406566215a6440fb27dca1f99337bb8 : SUCCESS in 1h 32m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2ea23e4885814f8891bc0b36f5219480 : SUCCESS in 2h 05m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/defa0013fa0143bbbc7c9d002faf6741 : SUCCESS in 1h 26m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5c32019a90fa4cafac705117344bde24 : SUCCESS in 1h 30m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7f07bb55f1524783a12cf32293e55dec : SUCCESS in 1h 35m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/8edb086d6cc04dc4afabd292e871ebac : FAILURE in 27m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e0ed0499b8e53617a5c711bf41bc1e18080c0aa0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-16 05:57:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a53dfbd97a8d4792344566804c7afa9c75d55aca","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-16 08:30:17.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91e7c1a11d9440631f69b183d2e057294c8b5efb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 08:31:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"30b081f4764fcb20cc9e7ac7f0e8439a227fa3e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 10:45:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ba252c5915446d7b88b3a7c87d6344d : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00226be9374740bc9baadb607de065f8 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/331d87faa2814ba9aedd5958c455e209 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8c60d5ec569421eab4e5060d17c6619 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/980ad7120aea479192b039320e91607d : SUCCESS in 11m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ce7ac2a1cb6145c585fc7f6bb5e75b7b : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/68e41ffc07e54c8f8f0b0bdb489524c4 : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/500d1e42d51f4cf98d54abf217255a37 : SUCCESS in 2h 10m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb8faa0b1af74747881af3808bf5c6b3 : SUCCESS in 1h 33m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c9ee5f33008e4a6f93671e27759ce8a3 : SUCCESS in 36m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9a4b6adbf91046deb07e87530275b096 : SUCCESS in 1h 03m 25s","accounts_in_message":[],"_revision_number":2},{"id":"11332ee193b191a4cd3586e0fd872eb826015896","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 10:45:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7e8870949dfbd148e20015281c31fbceb2ae2f30","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 10:47:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/295122159b274569a09af878ee48ecb1 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"399c2f17504fa156e65c664cc4bfeb1c98f7d9d0","revisions":{"01c00f438607576002139afeaf775f641f0a02b3":{"kind":"REWORK","_number":1,"created":"2022-06-08 14:48:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/42/845042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/845042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/845042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/845042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/845042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/845042/1"}}},"commit":{"parents":[{"commit":"b707808cad5783c58e2a2d6613fd9eec3dc3298f","subject":"Merge \"Remove unneeded sudo in lvs-masquerade.sh\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b707808cad5783c58e2a2d6613fd9eec3dc3298f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:45:51.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:48:07.000000000","tz":0},"subject":"Fix new pylint issues","message":"Fix new pylint issues\n\nThe new pylint release removed a few rule sets and added a check for generators. This patch updates Octavia for those changes.\n\nChange-Id: I2261342377e7acb1d2bc4fc742a351dfc66e4305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01c00f438607576002139afeaf775f641f0a02b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01c00f438607576002139afeaf775f641f0a02b3"}]},"branch":"refs/heads/stable/yoga"},"399c2f17504fa156e65c664cc4bfeb1c98f7d9d0":{"kind":"REWORK","_number":2,"created":"2022-06-08 15:27:19.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/42/845042/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/845042/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/845042/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/845042/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/845042/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/845042/2"}}},"commit":{"parents":[{"commit":"b707808cad5783c58e2a2d6613fd9eec3dc3298f","subject":"Merge \"Remove unneeded sudo in lvs-masquerade.sh\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b707808cad5783c58e2a2d6613fd9eec3dc3298f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:45:51.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 15:27:19.000000000","tz":0},"subject":"Fix new pylint issues","message":"Fix new pylint issues\n\nThe new pylint release removed a few rule sets and added a check for generators. This patch updates Octavia for those changes.\n\nChange-Id: I2261342377e7acb1d2bc4fc742a351dfc66e4305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/399c2f17504fa156e65c664cc4bfeb1c98f7d9d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/399c2f17504fa156e65c664cc4bfeb1c98f7d9d0"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
