)]}'
{"id":"openstack%2Foctavia~845043","triplet_id":"openstack%2Foctavia~stable%2Fxena~I2261342377e7acb1d2bc4fc742a351dfc66e4305","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-16 08:15:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2261342377e7acb1d2bc4fc742a351dfc66e4305","subject":"Fix new pylint issues","status":"MERGED","created":"2022-06-08 14:48:51.000000000","updated":"2022-06-16 08:17:31.000000000","submitted":"2022-06-16 08:15:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845043","cherry_pick_of_change":845042,"cherry_pick_of_patch_set":1,"meta_rev_id":"3357f4e0e9d648b36c9112a046c69c4538f64eb6","_number":845043,"virtual_id_number":845043,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-16 08:15:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-06-10 07:38:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-16 05:57:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-06-09 05:42:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-16 05:57:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-08 15:37:42.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-06-08 17:50:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-10 07:38:17.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-06-16 05:57:32.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"765e246cd973a61e0cdabe3a1f9c5e2f87c44bdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-08 14:48:51.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"48d74cc759f41570f8ec460a1f35ccdfad447412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-08 15:27:38.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c1b2f3e05c121ec1562989abd7448a479fde6979","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-08 15:37:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"581db0b994a8fe804084d07b4a46c747671915ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 17:50:09.000000000","message":"Patch Set 2: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f40c004d1c604cc2a624037320821d8e : SUCCESS in 11m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77eb5032747a4bfcaaff3eb2876fd38d : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9922b9adfe4499af07833275ef191d : FAILURE in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80350af7877c4360a7e0f9403fd1f99d : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e30f6671ef3f453caf1ea24afaabb007 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c22d8dfe564f451497e471ee2d8aa62d : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fd1434afb2a4014976c33a231e5040b : SUCCESS in 13m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b2769613998b48f688f50ede490a64d2 : SUCCESS in 5m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8631177eb1454b9e9582a4a187b9d220 : SUCCESS in 10m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cd24b5cfcb3348b2a1909e8e91b662bd : SUCCESS in 5m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/29229b595eda445cbdfa6c3abd1ae0f3 : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f5923794815e4c02b6c27a11e433d9a7 : SUCCESS in 1h 57m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21d818a204f54ea983dc349d0cfd302e : SUCCESS in 1h 59m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a2f47717800c46cc84f37810e2ceec25 : SUCCESS in 47m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2cf0d26555f640db92902687000850b7 : SUCCESS in 59m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ff4b48119a2a4fc8b8995bec171e4f38 : FAILURE in 35m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/56a46d22e641489ba3d1adc5178a07b1 : SUCCESS in 35m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a0b39689a63b4d3d95093447179a3127 : SUCCESS in 1h 27m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ed170e96acd94246abbeae9934d0294d : FAILURE in 2h 00m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3f21a35484a34f18a69568ca3098c431 : SUCCESS in 2h 07m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/04bac1fcf74841c8aec632315a22bb04 : SUCCESS in 1h 35m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d16807296b4d4218930577b4f041cff2 : SUCCESS in 1h 26m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8a4a41859c5248b19cc227339160ed38 : NODE_FAILURE in 0s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalived.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalivedlvs.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/loadbalancer.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/plug.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/server.py\n  Comments left for invalid file octavia/amphorae/backends/utils/interface.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/api/common/pagination.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v1/driver.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v2/driver.py\n  Comments left for invalid file octavia/api/drivers/driver_agent/driver_updater.py\n  Comments left for invalid file octavia/api/root_controller.py\n  Comments left for invalid file octavia/api/v2/controllers/__init__.py\n  Comments left for invalid file octavia/api/v2/controllers/amphora.py\n  Comments left for invalid file octavia/api/v2/controllers/base.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/api/v2/controllers/pool.py\n  Comments left for invalid file octavia/api/v2/controllers/provider.py\n  Comments left for invalid file octavia/api/v2/controllers/quotas.py\n  Comments left for invalid file octavia/cmd/status.py\n  Comments left for invalid file octavia/common/base_taskflow.py\n  Comments left for invalid file octavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n  Comments left for invalid file octavia/common/jinja/haproxy/split_listeners/jinja_cfg.py\n  Comments left for invalid file octavia/common/jinja/lvs/jinja_cfg.py\n  Comments left for invalid file octavia/compute/drivers/nova_driver.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/amphora_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/health_monitor_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/l7policy_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/l7rule_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/listener_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/load_balancer_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/member_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/pool_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/network_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/amphora_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/health_monitor_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/l7policy_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/l7rule_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/listener_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/load_balancer_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/member_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/pool_flows.py\n  Comments left for invalid file octavia/db/models.py\n  Comments left for invalid file octavia/db/repositories.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/network/drivers/neutron/base.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":2},{"id":"a374df28460cdf83d10f036674ad2e059f819b87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-08 20:37:42.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0cf265c3ef667e1a880264eeae7724529dfc1ca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 23:11:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fc062683e4247f092201af5c3f1241d : SUCCESS in 11m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be1839929a2b460da32fc2dbac92c9f1 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc55442b09014aa48fc7c6dae31c4539 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ec6f5288f224704928928d4707e2374 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8411af9e18f741fd8a89a8e4578c2f21 : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e69d977af5e54726abaa2b5d75f24b13 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85b7f036fec44479b69367fcb88563d0 : SUCCESS in 9m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a0ca3752d7ee47f99291de1b066b7931 : SUCCESS in 5m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7242edaf39fb468e877a9d1a3a2855ad : SUCCESS in 7m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0583c011483f43398c407ca8a7badc9e : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9f29fb626d5844baa7da304b4c7117a0 : SUCCESS in 8m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b54f27e1fd5542c5ac66eaaf1b84d5db : SUCCESS in 2h 05m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9da1bc21cb2a477b866b920a55e163bc : SUCCESS in 2h 06m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1fda8f8fb76141c096b178ae24a1ad22 : SUCCESS in 43m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/295f10a1497b4db99a7c9b477a9b18bc : SUCCESS in 1h 03m 31s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3198099774aa48e09dafd7bc91db299a : FAILURE in 35m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf37c6e5407440c3911be1a8e3e1299b : SUCCESS in 40m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a7c5896b9eff4f7db926947992bbe1f8 : SUCCESS in 1h 53m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9e638551951841d2904b471d9b8aa6d5 : SUCCESS in 2h 05m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7c9eb4b606cd474b9ce1d19894e44bee : SUCCESS in 1h 46m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/978c45fa67e34e2ebabad85e0e9ebd85 : SUCCESS in 2h 30m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/68dd8239ebbb4abeb95ff0b844d62a6b : FAILURE in 1h 53m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e223179810444e7fb5e5f732af2684f1 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2d0e98d149f97aa41017154271edea47361181b0","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-09 05:42:44.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f85afb21ee6999bc009ee4aa62fd266f3306d68e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-06-10 07:38:17.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0c77c30905aa2ab56c62fbd50e0fb01e2038c257","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-16 05:57:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"eefb7779356573aac314cf5af15ebb118d2651eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 05:58:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f16f028a846e1077f641de396de66d153b24047e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 08:15:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d661234a3bd4b91ac341ac7b9dd66d0 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10624438ac4b4b97af0bfe7e44d87d51 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c4d0c17430640bf981a20b81748cc84 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34905064eaf54484ac030297c69efa84 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e44f90499f44a5a60097ea57fc5bed : SUCCESS in 11m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1cbd1dfad65042ecad4bd58ce3d667a5 : SUCCESS in 4m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9438d94ae3254c2b85878b8f995e56f2 : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/38a2d1e75b054c3cbf008984392e8b27 : SUCCESS in 2h 10m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1000a99307be4ab086cae6d19eb673f3 : SUCCESS in 1h 43m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/36690ef85a4547b5b4d47cf4e50b9ca3 : SUCCESS in 35m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/48ab35ae82e74a59a53b3e594d65e0bc : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":3},{"id":"16981eed01b1689bc73e0469d7bda8c46b78396e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 08:15:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3357f4e0e9d648b36c9112a046c69c4538f64eb6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 08:17:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc01fa7754834d1598ceb9f2fbdf3ba7 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d668e1f2055e0124cce4a4b7abae06ade340e24a","revisions":{"541e1f9e7ce230035843c560d25d8bfc252bfa09":{"kind":"REWORK","_number":1,"created":"2022-06-08 14:48:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/43/845043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/845043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/845043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/845043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/845043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/845043/1"}}},"commit":{"parents":[{"commit":"74cd04d171874a708b340f61b6f19ad277bca56c","subject":"Merge \"Fix compile_amphora_details when using UDP listeners\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74cd04d171874a708b340f61b6f19ad277bca56c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:45:51.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:48:51.000000000","tz":0},"subject":"Fix new pylint issues","message":"Fix new pylint issues\n\nThe new pylint release removed a few rule sets and added a check for generators. This patch updates Octavia for those changes.\n\nChange-Id: I2261342377e7acb1d2bc4fc742a351dfc66e4305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/541e1f9e7ce230035843c560d25d8bfc252bfa09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/541e1f9e7ce230035843c560d25d8bfc252bfa09"}]},"branch":"refs/heads/stable/xena"},"27f828caa362c79184841d8d332a2e6df0dd4747":{"kind":"REWORK","_number":2,"created":"2022-06-08 15:27:38.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/43/845043/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/845043/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/845043/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/845043/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/845043/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/845043/2"}}},"commit":{"parents":[{"commit":"74cd04d171874a708b340f61b6f19ad277bca56c","subject":"Merge \"Fix compile_amphora_details when using UDP listeners\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74cd04d171874a708b340f61b6f19ad277bca56c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:45:51.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 15:27:38.000000000","tz":0},"subject":"Fix new pylint issues","message":"Fix new pylint issues\n\nThe new pylint release removed a few rule sets and added a check for generators. This patch updates Octavia for those changes.\n\nChange-Id: I2261342377e7acb1d2bc4fc742a351dfc66e4305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27f828caa362c79184841d8d332a2e6df0dd4747"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27f828caa362c79184841d8d332a2e6df0dd4747"}]},"branch":"refs/heads/stable/xena"},"d668e1f2055e0124cce4a4b7abae06ade340e24a":{"kind":"REWORK","_number":3,"created":"2022-06-08 20:37:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/43/845043/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/845043/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/845043/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/845043/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/845043/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/845043/3"}}},"commit":{"parents":[{"commit":"74cd04d171874a708b340f61b6f19ad277bca56c","subject":"Merge \"Fix compile_amphora_details when using UDP listeners\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74cd04d171874a708b340f61b6f19ad277bca56c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:45:51.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 20:37:42.000000000","tz":0},"subject":"Fix new pylint issues","message":"Fix new pylint issues\n\nThe new pylint release removed a few rule sets and added a check for generators. This patch updates Octavia for those changes.\n\nChange-Id: I2261342377e7acb1d2bc4fc742a351dfc66e4305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d668e1f2055e0124cce4a4b7abae06ade340e24a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d668e1f2055e0124cce4a4b7abae06ade340e24a"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
