)]}'
{"id":"openstack%2Foctavia~845044","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I2261342377e7acb1d2bc4fc742a351dfc66e4305","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-16 08:15:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2261342377e7acb1d2bc4fc742a351dfc66e4305","subject":"Fix new pylint issues","status":"MERGED","created":"2022-06-08 14:49:00.000000000","updated":"2022-06-16 08:17:20.000000000","submitted":"2022-06-16 08:15:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845044","cherry_pick_of_change":845043,"cherry_pick_of_patch_set":1,"meta_rev_id":"3651731f59c12d5180fe4a07fdabaf171c17f78c","_number":845044,"virtual_id_number":845044,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-16 08:15:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-06-10 07:38:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-16 05:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-06-09 05:42:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-16 05:57:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-08 15:37:05.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-06-08 17:15:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-10 07:38:11.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-06-16 05:57:38.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"90796de5726ea3a74344f236257d0b4b2f26e647","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-08 14:49:00.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"e79f2f71209fe3382a3bfbee7fe79f9e375223c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-08 15:27:54.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a8fdb18c9cd445f382d34aa655e0dcecf9846800","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-08 15:37:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"97b3a40e4c5b9743d42868661f12512b4c957ae9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 17:15:51.000000000","message":"Patch Set 2: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c26d0bed8ee3474db2fd506ec7e3c956 : SUCCESS in 8m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4daf2b1d5064bc1877be4f55852764f : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7bee4abe3b0402e9b16053cb58013c3 : FAILURE in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/026efc7078b449b6b633ab1b0720bb59 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/339af4f9307d48ff8b4fb30a50f1478e : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a08a4c68bf54bce983dfe04a68b42ad : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66060daa36384444aee7250099a49d43 : SUCCESS in 10m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/79e3b3519b2d4483b3289edc21a9045c : SUCCESS in 5m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d7b74d97209c441595487037bd3c833e : SUCCESS in 9m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3d63cbc2b83341a98728cc250df74c34 : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b31f2b64dbf148f7aa24ab22777e2ead : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0b9d83369e264567af26dbfbde0c456b : SUCCESS in 1h 41m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de99c7c2e599479793be9f9946adb0ef : SUCCESS in 1h 13m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39e90bfbd6a34549ac3233686b629c4c : SUCCESS in 34m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/47543ee51a6044ce9ff48636a843bb86 : SUCCESS in 32m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e52b303d27a4e859d3173ff0fbfafd4 : SUCCESS in 40m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a66d3cc90d7e4b5eba961dedb22fcffe : FAILURE in 23m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ec88bfc015ac416ca6110761a3eac5c0 : SUCCESS in 1h 28m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ebe7acd04a4043ddaec580860b494763 : FAILURE in 29m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a7aa4c178c2040e38e129df38432e45d : SUCCESS in 1h 30m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c87ca8272e30432aaf0451f4235d2eaf : SUCCESS in 1h 30m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/91b67cf8934145d291e4f55504588b35 : SUCCESS in 1h 34m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bf041dd5796c4fb5b62e8e73912a2097 : NODE_FAILURE in 0s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalived.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalivedlvs.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/loadbalancer.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/plug.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/server.py\n  Comments left for invalid file octavia/amphorae/backends/utils/interface.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/api/common/pagination.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v1/driver.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v2/driver.py\n  Comments left for invalid file octavia/api/drivers/driver_agent/driver_updater.py\n  Comments left for invalid file octavia/api/root_controller.py\n  Comments left for invalid file octavia/api/v2/controllers/__init__.py\n  Comments left for invalid file octavia/api/v2/controllers/amphora.py\n  Comments left for invalid file octavia/api/v2/controllers/base.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/api/v2/controllers/pool.py\n  Comments left for invalid file octavia/api/v2/controllers/provider.py\n  Comments left for invalid file octavia/api/v2/controllers/quotas.py\n  Comments left for invalid file octavia/cmd/status.py\n  Comments left for invalid file octavia/common/base_taskflow.py\n  Comments left for invalid file octavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n  Comments left for invalid file octavia/common/jinja/haproxy/split_listeners/jinja_cfg.py\n  Comments left for invalid file octavia/common/jinja/lvs/jinja_cfg.py\n  Comments left for invalid file octavia/compute/drivers/nova_driver.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/amphora_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/health_monitor_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/l7policy_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/l7rule_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/listener_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/load_balancer_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/member_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/pool_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/network_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/amphora_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/health_monitor_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/l7policy_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/l7rule_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/listener_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/load_balancer_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/member_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/pool_flows.py\n  Comments left for invalid file octavia/db/models.py\n  Comments left for invalid file octavia/db/repositories.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/network/drivers/neutron/base.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":2},{"id":"e5651971d7f62daa4450cb833fbd3b2fe5b337e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-08 20:37:03.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"6ef85fb02908c2819310ba358644d53052816f65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 22:12:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59795e46409047b3ae6874d38040f94b : SUCCESS in 8m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5f6fcd5555b4701b585fb3307e89a1f : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7597c7bab6349c884da82d07599e319 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0de7c6cecdf145f48e577401d752fd20 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3c240f46ab84be8b04404ad9ae32f2a : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/493baa4d6a114eee918d598b346615ad : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff985b4b57742e389472b2abd840e51 : SUCCESS in 10m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/90f1e25dcf4746bda429198c02205583 : SUCCESS in 5m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b03ea65a93254c5a8b9fc17022f5ab6d : SUCCESS in 6m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0518a838236d4c20b3334a735e0dfdeb : SUCCESS in 5m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d45e1c1a811a411aa48e3bce3163d2aa : SUCCESS in 6m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e44e4c063ef2444a8a93b9c8fbddf6e6 : SUCCESS in 1h 19m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cab445eef6e14ae6b7c292536f535c07 : SUCCESS in 1h 18m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6c3798ba1480472b854cbdb52e32ff8f : SUCCESS in 33m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b8c35ca2b6244bac92f94748005cd2ed : SUCCESS in 31m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11d9c69832004e2a9c3b394f1dab173e : SUCCESS in 33m 35s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0d327243c6304feb9fe41be0920222e8 : FAILURE in 19m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a292f1e61c4d486f94b7bd79ea4c0add : SUCCESS in 1h 34m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/02f7a25cf14e4c58840b427703199811 : POST_FAILURE in 55m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8d9adccee7474b14ad0d4eb8bb2ee15b : SUCCESS in 1h 27m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ea0e0f83ba534b3ca983bc3502e10100 : SUCCESS in 1h 30m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a493b3a054bf444e84021381fa648a42 : SUCCESS in 1h 29m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5e0ea80af48c48f49888dc0985ea7960 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bf1b5dc7ba89fdcde62ced367f48809b3b282856","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-09 05:42:57.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"36a42ad38c280d4d3441bc97db91e495bd126aa8","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-06-10 07:38:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"58575be798a85e2a3f41031555e9d96eea6cb5c4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-16 05:57:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"26facfa786f98f89c962e7c62ba8795596b7d106","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 05:58:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6b953cf03d0a3c6f8414fc538614c492280fb586","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 08:15:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b11beabf815439787262d97a94617cb : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50069f721401417d885d7934503e8314 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd3001dab1244a948af5c6206e0b3208 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e88946a6475142f084f7ee55614a2a1a : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/144dd46c27c74fc2938b355da460019e : SUCCESS in 11m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2726de0750044e5f85c0f8110a669d5a : SUCCESS in 4m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9a314e6f6787488a8ccbe4b39436cbaa : SUCCESS in 6m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b970c27a4dd0451a8ab83651f758a020 : SUCCESS in 1h 56m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6bafedfa071f459399fb83672b49542e : SUCCESS in 1h 12m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cf62f72806d04fc5b0dc916a630f99e1 : SUCCESS in 34m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/94890357664343dd81ce9b38f3d7149d : SUCCESS in 33m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c840cd7f827a48ecaed1729c3f2c659f : SUCCESS in 43m 19s","accounts_in_message":[],"_revision_number":3},{"id":"9218880bcb6b8d7f8aa0d1a5d17c49bf9c293404","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 08:15:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3651731f59c12d5180fe4a07fdabaf171c17f78c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 08:17:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce1f2586ee184793b0ecd58467337979 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a870525df503605b6e7c1ff932487ac61234dbb6","revisions":{"7819ba590ed6217e11a64a896e09a2e82635c32d":{"kind":"REWORK","_number":1,"created":"2022-06-08 14:49:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/44/845044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/845044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/845044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/845044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/845044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/845044/1"}}},"commit":{"parents":[{"commit":"bd250b93b74393f9d2699470288fa75e85d621c5","subject":"Merge \"Fix compile_amphora_details when using UDP listeners\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd250b93b74393f9d2699470288fa75e85d621c5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:45:51.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:49:00.000000000","tz":0},"subject":"Fix new pylint issues","message":"Fix new pylint issues\n\nThe new pylint release removed a few rule sets and added a check for generators. This patch updates Octavia for those changes.\n\nChange-Id: I2261342377e7acb1d2bc4fc742a351dfc66e4305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7819ba590ed6217e11a64a896e09a2e82635c32d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7819ba590ed6217e11a64a896e09a2e82635c32d"}]},"branch":"refs/heads/stable/wallaby"},"4596cd50c585ac6786c1ef1050d93fe7bf6b7d0c":{"kind":"REWORK","_number":2,"created":"2022-06-08 15:27:54.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/44/845044/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/845044/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/845044/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/845044/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/845044/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/845044/2"}}},"commit":{"parents":[{"commit":"bd250b93b74393f9d2699470288fa75e85d621c5","subject":"Merge \"Fix compile_amphora_details when using UDP listeners\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd250b93b74393f9d2699470288fa75e85d621c5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:45:51.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 15:27:54.000000000","tz":0},"subject":"Fix new pylint issues","message":"Fix new pylint issues\n\nThe new pylint release removed a few rule sets and added a check for generators. This patch updates Octavia for those changes.\n\nChange-Id: I2261342377e7acb1d2bc4fc742a351dfc66e4305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4596cd50c585ac6786c1ef1050d93fe7bf6b7d0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4596cd50c585ac6786c1ef1050d93fe7bf6b7d0c"}]},"branch":"refs/heads/stable/wallaby"},"a870525df503605b6e7c1ff932487ac61234dbb6":{"kind":"REWORK","_number":3,"created":"2022-06-08 20:37:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/44/845044/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/845044/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/845044/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/845044/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/845044/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/845044/3"}}},"commit":{"parents":[{"commit":"bd250b93b74393f9d2699470288fa75e85d621c5","subject":"Merge \"Fix compile_amphora_details when using UDP listeners\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd250b93b74393f9d2699470288fa75e85d621c5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:45:51.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 20:37:03.000000000","tz":0},"subject":"Fix new pylint issues","message":"Fix new pylint issues\n\nThe new pylint release removed a few rule sets and added a check for generators. This patch updates Octavia for those changes.\n\nChange-Id: I2261342377e7acb1d2bc4fc742a351dfc66e4305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a870525df503605b6e7c1ff932487ac61234dbb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a870525df503605b6e7c1ff932487ac61234dbb6"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
