)]}'
{"id":"openstack%2Foctavia~845093","triplet_id":"openstack%2Foctavia~master~I1633c2cacf1c4dc5c0aa605635545fae8085e296","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"7249":{"account":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"last_update":"2022-07-19 19:12:21.000000000","reason":"Change was submitted"},"10273":{"account":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"last_update":"2022-07-19 19:12:21.000000000","reason":"Change was submitted"},"33196":{"account":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"last_update":"2022-07-19 19:12:21.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-07-19 10:39:36.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I1633c2cacf1c4dc5c0aa605635545fae8085e296","subject":"Fix update/delete listener CA/CRL error","status":"MERGED","created":"2022-06-08 07:45:27.000000000","updated":"2022-07-19 19:13:41.000000000","submitted":"2022-07-19 19:12:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845093","meta_rev_id":"1df96940d9ea0230cf3cae9aba28b5c7654c1690","_number":845093,"virtual_id_number":845093,"owner":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-19 19:12:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-19 15:02:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},{"value":2,"date":"2022-07-19 16:59:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-07-19 10:39:36.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},{"value":1,"date":"2022-07-19 16:59:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},{"value":2,"date":"2022-07-19 16:59:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-07-19 10:39:36.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-08 08:11:31.000000000","updated_by":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-06-08 08:11:31.000000000","updated_by":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2022-06-08 10:16:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-09 05:57:18.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2022-06-21 02:47:46.000000000","updated_by":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-07-19 16:59:24.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"b520ec5ef774543143f4413ee078323ef32b302d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-06-08 07:45:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a8493be1bf695126fad48ca50d21abb0f9c95e7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-06-08 08:43:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b9316189ed7bffe5f14c4bfb99fcd61953f2402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 10:16:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c79cd2a1ce054435b600e463c382c2c8 : SUCCESS in 12m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/caba2f2a919146ff8b83bb1b26e3c705 : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c065adafd5d40939ea930cf69e75d80 : FAILURE in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18957bcc4e4e448b8557f79463ef89bd : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98bd1d9bb35f4a8fab8a8274b7d23eb9 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71b7d65e037c4c50befb4d54ba0c7634 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/826fbd4d57214fbaad1f5c1abdb7e260 : SUCCESS in 11m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/af0d3384a5b540b7891a87824951c231 : SUCCESS in 5m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ac04a99e675d40898d8c8af32784edbf : SUCCESS in 9m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/da4bfbb0d5b9468ea1f05bb04f757a3e : SUCCESS in 5m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b59ce321c76d4a02a28f9ca0e4548f58 : SUCCESS in 8m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c1affe5c7f214700b345c8b3f59384c7 : SUCCESS in 8m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/170d69398827466cb9a105ed0c476244 : SUCCESS in 10m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/af8302e05c05427e97d4917ade1bca47 : SUCCESS in 2h 10m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f457bc8e6de4dd8b4fa92d61f5227ff : SUCCESS in 1h 36m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8c248e9d50974d1db669af1179f7fd19 : SUCCESS in 42m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/27e43fad87be4b6bb1ebdbaac696e86b : SUCCESS in 1h 17m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/12b2499d6c664e47aaa3079486d2fa3d : FAILURE in 13m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7643c2940e25478295a59efc6e81a86e : FAILURE in 18m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0cd704d9648d480d87bb6e9c0a11b56e : SUCCESS in 2h 09m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0f855bf5e5234d8892535d81555d30ff : SUCCESS in 2h 08m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5010b481223f4971aafcabffe92fd4d3 : SUCCESS in 1h 36m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d187a42dc9fe4486b17b1e571287b658 : SUCCESS in 1h 35m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d1bfc040bd9f4f4399daf40faea7641c : SUCCESS in 1h 47m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f7de27f3ab149839589953b87851531 : FAILURE in 14m 36s (non-voting)\n\nWarning:\n  Comments left for invalid file .pylintrc\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py","accounts_in_message":[],"_revision_number":1},{"id":"f56e0c137f6f273d19f3996acfc04e332bb43cf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-06-09 05:46:42.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"54e1cb2d171f189a0a14d9496ea1843c324569ac","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-06-09 05:48:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b7e4c987ac8eb8a74aac1883637f209963ce45d7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-09 05:57:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5aa9eaa00c9d1d4f100f8d43a39de6e560bafe8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-06-09 06:40:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1798ef59c86f6055c95e7c6b437eb232d6c9a9c4","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-06-09 06:47:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8424dfb1f8214d2b7a8d1f260b6f438f0c5575ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-06-09 08:18:39.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"153a43c24d703c783a2a9ac36d63b2b5d593009e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 10:48:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/788d750b321647ddab06b7756cf09582 : SUCCESS in 9m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90c07707237f4181af11fcad8147d665 : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca8c91fc3a554341b27102cf49b16725 : FAILURE in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbf8915c728347168732520e1f07efa9 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5afc91ef6b974bf29b9fd48c954ee1d2 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbc5ade5780b4ea9b1459fb18a0fb93b : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d03b9e76e114470aabc31a65785c5ee : FAILURE in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f842674a3ac84608b5410d1da2f27441 : SUCCESS in 5m 12s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/74a22a05bbfd42b692e7ac5f025cadf2 : SUCCESS in 5m 43s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f561e6e7e4174834aaa244e31a71f62a : SUCCESS in 10m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2af99075daef40719992befd5332aa56 : SUCCESS in 6m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/61334a94a2f1447bbbce190329b7737f : SUCCESS in 8m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/78fccf0052ea4b03917c97d6e822937c : SUCCESS in 9m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d59592022c874d59a609503abdc33bf0 : SUCCESS in 8m 05s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/507a8a7a39e24110a6207abfcaac1415 : SUCCESS in 2h 24m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d6f59a9c7ae547b6bd883bb3da723b88 : SUCCESS in 1h 35m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3f683e1df48e4e4384682f860c326c95 : SUCCESS in 36m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e7cd1adfb0a14328bd06e7bd5d28ad51 : SUCCESS in 50m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4c56f8d1f8ab4a0aa581ca4592fe04e4 : FAILURE in 15m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/431a0d8b3ee947d2b9b209a753e256a7 : SUCCESS in 35m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c6a135d669f44c49a30040c8e3c7e0ae : SUCCESS in 1h 25m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/649b53ff15f14f4ea5a3e71a74305bfc : SUCCESS in 1h 55m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/109cee27f6ac4ae895002b51f4efbd8e : SUCCESS in 1h 35m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/15794a423322420ea97050e5bd51ecb4 : SUCCESS in 1h 45m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ef982fe36d6f4d4d804d9e38e6424eef : SUCCESS in 1h 26m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/10f7e835ebf8485a9a4d0fb5e57eea5e : FAILURE in 12m 48s (non-voting)\n\nWarning:\n  Comments left for invalid file .pylintrc\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py","accounts_in_message":[],"_revision_number":4},{"id":"3709387eae3c3f12c0bf78853927220fa63a8a5c","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-06-16 06:42:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"04f4bfa8083d9843a9cfbb0b1b310b458dfb2a30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-06-16 09:01:08.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"871f37a5c094525193bd418b375dce4c8cc7f2a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 11:21:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9e5a7f370be4592937c51201f017662 : SUCCESS in 8m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce49911ccbcf4670a58d940f7c4fe6df : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c672366e4bc54239aeeff68494c6796b : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b4d30938e3a464f9edf47eb288af3a1 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27c4d865865e4c1aa78a4f58b1e37ce6 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ac1e6befbf44894b9c6f4b8394f0d05 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eec0df07c5243559bb8e58d40e2db9d : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29deb1b9d4fe44f0bfc0508df80406ea : SUCCESS in 5m 41s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/458a9c4753e14a2e8320af8e44eb518e : SUCCESS in 6m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/06bcf566ca0948cfaa9cf6a3fe4ec941 : SUCCESS in 8m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/af1489e63b654861ba2d19c3ba7ead1b : SUCCESS in 6m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c5e99566d6b643ce8739c2a68b1edd5b : SUCCESS in 8m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/196c1f0eca1f453b93a0222a021c7e5a : SUCCESS in 8m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/abc0bc38f3594a76a7c94a86fd5184b0 : SUCCESS in 8m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/08fb83060dd346f496afca69a16a02b5 : SUCCESS in 2h 14m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb3202755393430b9e671cef7d7a95e7 : SUCCESS in 1h 30m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7794184f354a4971b79c2f7a4472d903 : SUCCESS in 37m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/66d7a375cea34d16bcd03285eea8c87a : SUCCESS in 53m 22s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fe03f4d9d5934a5893a5480037eb11dd : FAILURE in 12m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ae9cc3598a041dd9d286cfb7724e005 : SUCCESS in 36m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8407750082c846629b93cace048f0032 : SUCCESS in 1h 42m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8f5f1950c0f248529811c7373090137e : SUCCESS in 2h 00m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1125425b4816425f96c7b6d90a8f78bd : SUCCESS in 1h 40m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a13242abbfda41189b619620dd073ad9 : SUCCESS in 1h 27m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/27c9740753cc4c2480f102b56c69cb6e : SUCCESS in 1h 46m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/bbe5585164c642a6920cf2e5ac2e9562 : FAILURE in 9m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"69063092579bfcde145295dc6e645ef2d3892d3b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-21 10:44:40.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"187b02d225340099f4873b438948ef38c6e35734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-07-18 09:33:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6b2a2ddabe670be5396355046b1ec4c90414e6fa","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-07-18 09:34:56.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b8275c3bee0a88a438f84d8ed18328d8a93f64ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 12:49:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3339b0747c6f4502b540f1ad1aa114c4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fe08e7e02e344f0ab7dc81d19ba71ed : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872d45df87bc41fe85ab17016001bffa : SUCCESS in 9m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/825a57ab0999492aaac3cc055cb3111d : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48fde22820b94e4cbbb2d3db15d9638b : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d96c3dba3054363a58d39c1618c4a25 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d856294b5cc49ffa91ed28ade16b3c1 : SUCCESS in 11m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da91212c1e2a491e82b82d32f1ed3cc3 : SUCCESS in 5m 38s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7ab90f73de08472e9f7c7e97851142d3 : SUCCESS in 6m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3e0ac4b88e5c47a499c029ac368cd052 : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2997df8e9a5d4a8aaed8f08a7e3e72ca : SUCCESS in 5m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ca94c4ca6c5c44bebcab3d0c0e324337 : SUCCESS in 7m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/859339d06e3a42bd9e3f7eb4713b9efd : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f4346c7de8b746e4b11eb29aa8257b91 : SUCCESS in 8m 00s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c63a28d1f4564e0bac91cc2fb44f9c0c : SUCCESS in 2h 13m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/143d5777b2d24d46963fe355176dbade : TIMED_OUT in 2h 42m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5d302f55ccfa46b29e1e4949ee79e026 : FAILURE in 1h 13m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d851a9d68f4d4d9884732fdeacc8e9f2 : SUCCESS in 1h 24m 17s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e6953a7617404692bc01a83db3354905 : RETRY_LIMIT in 15m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69355def53eb44b985bdc1fefd35fdbe : SUCCESS in 57m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/82a4717beb9f405faf26d1a04f2245c9 : TIMED_OUT in 2h 42m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0fd6fd85e6d24509a00618d8e173f554 : TIMED_OUT in 3h 12m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/db97609667e7465c933daf45608db943 : TIMED_OUT in 2h 42m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e212bad99f5545de9e7b0547123d3214 : TIMED_OUT in 2h 36m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fdc9d894ef174383994086b1dff65287 : TIMED_OUT in 2h 36m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ef819582cfb442baeccce04fae94d14 : TIMED_OUT in 2h 47m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"901e4e3deb5e1af8e96d219c2b21a134a5df969d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-07-19 06:20:18.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"050a656e93a77544c577b26c116ebdc947007e20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 08:55:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b31519ab94c412f88412f232eeba133\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc3e979facc046758b9b70ff93280316 : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19c8bbe441b74511ad0cf4613d35e900 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e3e9f0c697249278f4251518cc3ccbf : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71c313de997947dd960b6b4affed69a9 : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fff7f0e2a6814f63aa6faf0978d01230 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2827e4c0bca449ea9abb0f789db818e1 : SUCCESS in 11m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef59f43776bd4d58b9079c02bcb8cb57 : SUCCESS in 5m 34s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/daace6baf70e4828979d3c390884ec8a : SUCCESS in 6m 20s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/170e1b351ba843929a41f9c587e187b3 : SUCCESS in 8m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/00fa46bcc97e41a78ac20e7ec8897061 : SUCCESS in 5m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9626ff9209a74e609e21b5ec220de73b : SUCCESS in 8m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ae8dfdf07187452d98f88691cd0beb1f : SUCCESS in 8m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ce4b6d1410f84a5b8a66b7123edbc730 : SUCCESS in 9m 32s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b5fcb0c7926140b3ace2efb6bc4888c8 : SUCCESS in 2h 25m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7fbfc88230f948c097ea2d359d124e52 : SUCCESS in 1h 26m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/58b2998af3864f2886654d9b09738544 : SUCCESS in 44m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/994c32fc47e44ba08a83944a3c473dd4 : SUCCESS in 57m 35s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6eb3990e3e1e45e783397e2103982f29 : RETRY_LIMIT in 14m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9fddd5135ada4fec8cbe4d69608a88e3 : SUCCESS in 40m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2046c29107fd48e481c91928ebbbb836 : SUCCESS in 1h 49m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d6f67f9c4fa54c9f8dc9fb4769e84a90 : SUCCESS in 2h 07m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7f240e02e5434d0baea1bd546c6bebb0 : SUCCESS in 1h 54m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/29c735f3e4f74c47a56c4e668acc72fd : SUCCESS in 1h 51m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9238e788d4a1494f8254307ab3fbe58f : SUCCESS in 1h 51m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a08d00186d84fc3a0fc46c2dfffb936 : SUCCESS in 2h 33m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4baa6d2911ad756631f18ebd070e77547ec5ca54","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-19 10:39:36.000000000","message":"Patch Set 7: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"b200c4e2580dd1a3d9b585e08188543833251731","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-07-19 15:02:46.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"982374ad2167cf75dc4e5648a49c63fdc7b98a8f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-07-19 16:59:24.000000000","message":"Patch Set 7: Backport-Candidate+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8dccaa018c863b8acadf3ee235123aedb479d04b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 17:00:04.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"77e760435aa357f2107d2bacac58a59729b5110b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 19:12:20.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dc7268eb8d945babff8dd2db9fecc15\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e09bacc2018d45a995bc5d4f3e5079b2 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68d50189ef4e494ba7ba6b5e9a96f535 : SUCCESS in 8m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7207fc1a29ea44bc930b947ca0e253cb : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf0e8314d0934b059b634f1be7fb13d7 : SUCCESS in 15m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a86eabe783a643caba0cd1c89e99cc0a : SUCCESS in 8m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ee81c4623a244b398f4a3524cdfd5ee2 : SUCCESS in 7m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c00c8ca9f8b64f589061c8c94258afe5 : SUCCESS in 10m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/08c1ba39546c4a0a86411dd04e74e2a5 : SUCCESS in 1h 49m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f63f700822164c458d11ce9547643798 : SUCCESS in 2h 10m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8d09ccc7d6284a3781fa7652328f2b9f : SUCCESS in 38m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e29e889b99ae4ad887afc828a74ad99b : SUCCESS in 1h 29m 14s","accounts_in_message":[],"_revision_number":7},{"id":"c7e95bb3e9ed688997c075a1293afa8904d7bbbd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 19:12:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"1df96940d9ea0230cf3cae9aba28b5c7654c1690","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 19:13:41.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fdf90178597487c8c0947d313bc7e52\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3193f8839e74359bef90b1fe4d643a1 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed196f3030bb4e0393671131cc5f6ce5 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"9a5273d3ea91109c8f22671fd7215eb7b953b975","revisions":{"e77aec1b302be3b20615b009a5ee7ad957340c9d":{"kind":"REWORK","_number":1,"created":"2022-06-08 07:45:27.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/93/845093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/845093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/845093/1"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-06-08 07:43:56.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-06-08 07:45:22.000000000","tz":480},"subject":"Fix update/delete listener ca/crl error","message":"Fix update/delete listener ca/crl error\n\nChange-Id: I1633c2cacf1c4dc5c0aa605635545fae8085e296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e77aec1b302be3b20615b009a5ee7ad957340c9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e77aec1b302be3b20615b009a5ee7ad957340c9d"}]},"branch":"refs/heads/master"},"d7a167a1261563b64ce48da306088a5a4ca23f83":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-06-09 05:46:42.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/93/845093/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/845093/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/845093/2"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-06-08 07:43:56.000000000","tz":480},"committer":{"name":"conna","email":"yang.shaoxue@99cloud.net","date":"2022-06-09 05:46:42.000000000","tz":0},"subject":"Fix update/delete listener ca/crl error","message":"Fix update/delete listener ca/crl error\n\nStory 2010081\nTask 45577\n\nChange-Id: I1633c2cacf1c4dc5c0aa605635545fae8085e296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7a167a1261563b64ce48da306088a5a4ca23f83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7a167a1261563b64ce48da306088a5a4ca23f83"}]},"branch":"refs/heads/master","description":"Edit commit message"},"692a514b55fd89bfe7a32e67c556b7738b2d5f11":{"kind":"REWORK","_number":3,"created":"2022-06-09 06:40:41.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/93/845093/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/845093/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/845093/3"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-06-08 07:43:56.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-06-09 06:37:22.000000000","tz":480},"subject":"Fix update/delete listener ca/crl error","message":"Fix update/delete listener ca/crl error\n\nChange-Id: I1633c2cacf1c4dc5c0aa605635545fae8085e296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/692a514b55fd89bfe7a32e67c556b7738b2d5f11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/692a514b55fd89bfe7a32e67c556b7738b2d5f11"}]},"branch":"refs/heads/master"},"be655522f5f4cc5199f93697b38a481589da4d50":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-06-09 08:18:39.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/93/845093/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/845093/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/845093/4"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-06-08 07:43:56.000000000","tz":480},"committer":{"name":"conna","email":"yang.shaoxue@99cloud.net","date":"2022-06-09 08:18:39.000000000","tz":0},"subject":"Fix update/delete listener ca/crl error","message":"Fix update/delete listener ca/crl error\n\nStory 2010081\nTask 45577\n\nChange-Id: I1633c2cacf1c4dc5c0aa605635545fae8085e296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be655522f5f4cc5199f93697b38a481589da4d50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be655522f5f4cc5199f93697b38a481589da4d50"}]},"branch":"refs/heads/master","description":"Edit commit message"},"565b072902d8cca467aed2f382a75f5c041cd53c":{"kind":"NO_CHANGE","_number":5,"created":"2022-06-16 09:01:08.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/93/845093/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/845093/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/845093/5"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-06-08 07:43:56.000000000","tz":480},"committer":{"name":"conna","email":"yang.shaoxue@99cloud.net","date":"2022-06-16 09:01:08.000000000","tz":0},"subject":"Fix update/delete listener ca/crl error","message":"Fix update/delete listener ca/crl error\n\nStory 2010081\nTask 45577\n\nChange-Id: I1633c2cacf1c4dc5c0aa605635545fae8085e296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/565b072902d8cca467aed2f382a75f5c041cd53c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/565b072902d8cca467aed2f382a75f5c041cd53c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b86e5aa1265c63f1d18753ed349294760a05c75c":{"kind":"REWORK","_number":6,"created":"2022-07-18 09:33:49.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/93/845093/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/845093/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/845093/6"}}},"commit":{"parents":[{"commit":"65b1c341e209f34593e87ffae04478c71bf66330","subject":"Merge \"Ignore status update on deleted objects in driver-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/65b1c341e209f34593e87ffae04478c71bf66330"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-06-08 07:43:56.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-07-18 09:33:44.000000000","tz":480},"subject":"Fix update/delete listener ca/crl error","message":"Fix update/delete listener ca/crl error\n\nFixed \"Could not retrieve certificate\" error when updating/deleting the client_ca_tls_container_ref field of a listener after a CA/CRL was deleted.\n\nStory 2010081\nTask 45577\n\nChange-Id: I1633c2cacf1c4dc5c0aa605635545fae8085e296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b86e5aa1265c63f1d18753ed349294760a05c75c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b86e5aa1265c63f1d18753ed349294760a05c75c"}]},"branch":"refs/heads/master"},"9a5273d3ea91109c8f22671fd7215eb7b953b975":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2022-07-19 06:20:18.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/93/845093/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/845093/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/845093/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/845093/7"}}},"commit":{"parents":[{"commit":"65b1c341e209f34593e87ffae04478c71bf66330","subject":"Merge \"Ignore status update on deleted objects in driver-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/65b1c341e209f34593e87ffae04478c71bf66330"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-06-08 07:43:56.000000000","tz":480},"committer":{"name":"conna","email":"yang.shaoxue@99cloud.net","date":"2022-07-19 06:20:18.000000000","tz":0},"subject":"Fix update/delete listener CA/CRL error","message":"Fix update/delete listener CA/CRL error\n\nFixed \"Could not retrieve certificate\" error when updating/deleting the client_ca_tls_container_ref field of a listener after a CA/CRL was deleted.\n\nStory 2010081\nTask 45577\n\nChange-Id: I1633c2cacf1c4dc5c0aa605635545fae8085e296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a5273d3ea91109c8f22671fd7215eb7b953b975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a5273d3ea91109c8f22671fd7215eb7b953b975"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[]}
