)]}'
{"id":"openstack%2Foctavia~845146","triplet_id":"openstack%2Foctavia~master~I2261342377e7acb1d2bc4fc742a351dfc66e4305","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-16 04:49:43.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-06-16 04:49:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2261342377e7acb1d2bc4fc742a351dfc66e4305","subject":"Fix new pylint issues","status":"MERGED","created":"2022-06-08 14:47:28.000000000","updated":"2022-06-16 04:51:14.000000000","submitted":"2022-06-16 04:49:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845146","meta_rev_id":"ebd801342f231d859e31736e6585ee797bc34507","_number":845146,"virtual_id_number":845146,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-16 04:49:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-10 12:12:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-06-15 12:20:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-15 12:40:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-08 14:56:22.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-06-08 17:45:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-10 07:38:02.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-06-10 12:12:45.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-06-15 15:39:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"abc6493590539f2243a2a8d729f22f8d3de85c43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-08 14:47:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee36741b278b90698c2a3dd6941bc136dcd1b7b5","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-08 14:56:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"932bede681f8fb36d83d55d670f86f58fea2c490","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-08 15:27:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"735ae117db2fdf525df2dfb020efe42ce572ef98","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-08 15:35:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"dd4f1c657f0e769964531ad01af663e81384564f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 17:45:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fecc7037ea4e450fa4d5c037eefab10d : SUCCESS in 12m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a59a65f8970545d5bf6d4ca55a330e80 : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b77bf1517e84dc48ca2824e0af95a7b : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c67020506bf94fa984ae1ef8668f73e6 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b61b7e3991b4dd2ad64cb55b5243574 : SUCCESS in 6m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2674cab3f1af469ca5b8cdf0a1030cf7 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c70cefb1e934086a572296eb2975d28 : SUCCESS in 14m 34s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/adc105e8adc447bcbfde9e300870b986 : SUCCESS in 5m 51s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/337c2f88430640d48d672aeaf8468182 : FAILURE in 12m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2425a9f0a3a5485baf2f8db063d7b44e : SUCCESS in 5m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a19f352f1dca456889b7275808ffd969 : SUCCESS in 7m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9ea3177a2e754767ad3685f3d525696d : FAILURE in 10m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4148df581c884737ae5467367d41082b : SUCCESS in 7m 55s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aba1f20ade6544d08fca24e36cdd1b37 : SUCCESS in 2h 01m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7479a3643ac24101b837e7deb0d6cfe5 : SUCCESS in 1h 24m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1544d44fcf894443b33e2263c2de1b53 : SUCCESS in 39m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/393687f4bea04ccda43539baf0d70f78 : SUCCESS in 57m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b181971ee24f49879825115efedc0d72 : FAILURE in 12m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e508c35518f49aca7474202c6c90cfb : SUCCESS in 33m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8e5a206cc16b46d1971db13a0058ff6b : SUCCESS in 2h 05m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/15eacffc3d014418a34512cbd9f4b150 : SUCCESS in 1h 50m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c2794ef1862346c4bf2e6333de23acb2 : SUCCESS in 1h 45m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8fdbb08b8ad846bb9f37cdcca579d055 : SUCCESS in 2h 16m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3c8a5e7f106f4325978435a1b7bc1ff4 : SUCCESS in 1h 48m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/90524bdffe55412981940930eed69fa3 : FAILURE in 12m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89d2f038d330359c8c27b31ddd2673d5d63b8f26","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-08 20:22:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"10406d8e684858dc46a774468818bc07d23a3b98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 22:38:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ec28b94f86445c29f75854fecefd4b5 : SUCCESS in 10m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82b01b796e58439fb6533717fe942458 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62ecb6b2fdbe4e96aea6efee90e91186 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c28e95ffa17c422bbc95a75e90db6dd5 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1df6df7aef184108b583ac3a866e1cd1 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7195d6849e04c18a37ff7ff99492f13 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/732577b254f84c878da54d28df1db1c0 : FAILURE in 4m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fd8b6e2ff3274485b3e1554192768334 : SUCCESS in 5m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f711978870df4931a77cc6494651e87f : SUCCESS in 7m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fd111f1b71284f7e9cb65f35605a1494 : SUCCESS in 5m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/64385d37cc55455ba5f6777e2189f0e0 : SUCCESS in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/083a2be7370e4ea59d95b0284763db47 : SUCCESS in 7m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/60186886e20543a38f580569881541b0 : SUCCESS in 7m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5f983df521b440a395113ee315a389b8 : SUCCESS in 2h 03m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a67cbae253b54a46b803a029140e6c4c : SUCCESS in 1h 50m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b5786f32cb374f65845491f8309b6d99 : SUCCESS in 36m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0e819ee3017c4c14bd96da179e128607 : SUCCESS in 1h 07m 07s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a8ff6a72e9264c5280abe83147738a4c : FAILURE in 13m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c1295839e4b4985a82270e93f4e6100 : SUCCESS in 35m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f03591eb33914dc2b92d764969584752 : SUCCESS in 1h 38m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4b971677b89a4dba879e118626dde1ea : SUCCESS in 2h 12m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e6a8d97e768a45b795fbc9fd3e981aba : FAILURE in 16m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4ad5de8edcda4a8689ca7d20c36ebe26 : SUCCESS in 1h 35m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/75e71cc5ad3e4ea08d72d2433bb2d9d1 : SUCCESS in 1h 54m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/1da1534d3fe9496ba16d5a9a0a9717bc : FAILURE in 12m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5dca7e3002df96131521f88f777d592da9517743","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-09 05:41:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"274a46668ae29ae7361d2b9c0152b11d738bb627","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 08:02:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c707d01fdbf84909bd690fa53b08adb3 : SUCCESS in 12m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e780a810c034fbb9456549c6441d6ef : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f1434dcc08c43caa7d587796dc3ca64 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0775f94052014b91b387e689017937b5 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6337b15e64e94563a8acaa90cc88a255 : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c0c812eb0954fb6bdcb174e41e7bb64 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1412f487bd4c42b297cbcd15b80dc972 : FAILURE in 4m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b0e74365835f4dbe9766c8cf4d058795 : SUCCESS in 6m 28s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f0707922439946fea9ba0058663e9288 : SUCCESS in 7m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8f6e25d9b91f4fd492822424f56aba4b : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/017ac4e1f3d34276a68ad80459f62ea4 : SUCCESS in 8m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cdd8e6584e0248cbb439df11178be317 : SUCCESS in 9m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/22a4e7064bee49d4beb5f189588b390b : SUCCESS in 6m 50s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0a00a0624cfb4a1ea60e7202417faf80 : SUCCESS in 1h 49m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4763378ee6894ec3a39fbd2382341466 : SUCCESS in 1h 26m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c919e38910b144b5ab078c9aa370869d : SUCCESS in 40m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e9630d0b242a4e52988b67c8d9a53d05 : SUCCESS in 51m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9a7e9a9752ab4a7c9819bc5cdf8a72d6 : FAILURE in 17m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/952b71d9651f4369bae5b5799ca01a53 : SUCCESS in 32m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c27b8c6bef344f868efc07aa2d9dcd04 : SUCCESS in 1h 23m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f605efa2e7e0489694a819375b90cee2 : SUCCESS in 1h 55m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cd1757bdce09433f868bf0435dc656cd : SUCCESS in 2h 08m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/272dc17bcc3d477ebc64112544c02844 : SUCCESS in 1h 29m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b1872a4317a14f8ba8ffa45ef923418a : SUCCESS in 1h 56m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/f1be5cd4a5aa42dbb71d3a6453a7b26b : FAILURE in 12m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d14a143b6bd498a1754f5c99b510c00a21ae1420","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-06-10 07:38:02.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"195786559ca36c13c91d21d417ed6ce28dcc65fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-10 08:38:46.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"dde2416eaaf8ec11f446d32554c1fc379a614c77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 09:55:46.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"4cbb171a1bf7a0dcdaa7ac1f3720c9b0cb76df38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-10 10:20:08.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"da2ae37d65ea634b8598a035ac06205b5104de55","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-10 12:12:45.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"263445f19a7ad68fce2eed34e130b595f0a9d7ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 12:39:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db4d7e3a26344a8cb039f1a018f87bff : SUCCESS in 9m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/516cde61e1cb4f4a9fb3d28627660633 : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c978f68a4f944b6ebcbbdba2ae367cde : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5020643618a46a4b6a2d3a7f8ed6100 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a800dca37c84fb396fe257c6dd80ef0 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b61e17202c84c82b64cc3dbcfaf9c60 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59f8cca4ce054449879730eaab798e14 : FAILURE in 5m 36s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ef78442384cf47749a1f3aced5216a82 : SUCCESS in 5m 12s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/398f0a8c2362409f84e7b2f5c01e10c8 : SUCCESS in 8m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a8b5694919104c30b0250107323f6051 : SUCCESS in 5m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/41af344932094ebaac7a1f7f5d5687ab : SUCCESS in 7m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0288bb5bfb9a4e7089c0ba157038d1f8 : SUCCESS in 9m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cbb7d11047d246d390978d0aa826736e : SUCCESS in 8m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5449cb55ca72439cb576c07e37f68d3f : SUCCESS in 1h 50m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/300e66fd661a41d2a40004603dfc2673 : SUCCESS in 1h 14m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f84016fe9e994448b53438e1eedc7092 : SUCCESS in 36m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/db4f8fe587df40f18fa2542619ed89d1 : SUCCESS in 51m 15s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5405b6a3268943c382e1f28ba0ea8654 : FAILURE in 16m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d552f65f1d1442ab5c874ae2c19abbe : SUCCESS in 35m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fd694688364d464fb2269b6cf57764fa : SUCCESS in 1h 37m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d15c8ba3b77240eba0aa37b4c45c2718 : SUCCESS in 1h 57m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cfe7541ecbd7419d85ec4fa9f1d39242 : SUCCESS in 1h 33m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/39e942a00b09474e816821b024bd8453 : SUCCESS in 2h 17m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1dad8f0c677043c8b73539fda7f18580 : SUCCESS in 1h 30m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/8cc5ca802be248ab8a47f1b9cfabfa30 : FAILURE in 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c660a0ed2c185f6d4688a08052ed0784d1b55ae2","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-15 12:20:01.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"42a578e59b396b7d9436ab91fca23918e8fd8fb7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-15 12:40:08.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b40f6b49cbc3b149a518c11661be3358eb6c2416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 14:38:49.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0af3864dffe845b99707ca10fca86987 : SUCCESS in 12m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e19eaa54acd94f2bbb49c31d9aa8d778 : SUCCESS in 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f6bf8ae7ea949cca207bc496bbd7fdf : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26bc1b84e57f409c8bafd0bf207429a8 : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c2b2772f0424c1291ab2179596667de : SUCCESS in 7m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f63bd347662b4ad38e159e4078a0b47e : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aabb54c8dc00495aab1f9b921cf5f2fd : FAILURE in 5m 48s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ef026a8628ab4d2dab7af8a3b4c9c721 : SUCCESS in 6m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a2e77a8139154bdeba89e975f4d40e3c : SUCCESS in 10m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/818fd97293134ca0a16f44803ef9ab56 : SUCCESS in 6m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1c423b27e138433680adf2155cc3f643 : SUCCESS in 8m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9cfcc7cf38f543be9bde6268f1416703 : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/97bb73ae76d545a49bdb02e9aa7dcb97 : SUCCESS in 8m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c7e04727d5b54258a52c2183115cfc81 : SUCCESS in 2h 13m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/725ee86fe30948c7879ea746b70ac156 : SUCCESS in 1h 38m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ceea30aae0c04e09a9313bad8950c0a4 : SUCCESS in 34m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c10ce4c84f94441ea10f38df366fcd99 : SUCCESS in 50m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f0f51d8163ae46e9b90937d694a866c0 : FAILURE in 17m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d6d96ef96ca4ab68ae86c9cc433e2ae : SUCCESS in 33m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/418e915edf42457ab80c4e6305266667 : SUCCESS in 1h 26m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3db5da2837f9463ba2cdad902d976852 : SUCCESS in 1h 45m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5df0067631e6451ebac43ee8376243c9 : SUCCESS in 1h 32m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ff6ea9cd4fec4a44916f190d8d354a40 : FAILURE in 2h 01m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f089548c79d345f79fa2d510ef95ed4b : SUCCESS in 1h 48m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/97635e400fd64fb4a0dccc7340db4bb4 : FAILURE in 13m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c8ab0b739c0d6b53453315a15d129f4b7657c19d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-15 15:39:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2e4571ad6b17ea415b72ceb38b971c3e830cd9a8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-15 16:03:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c91b00e1e3d5539efd42224dc7923f3cc8f98dea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 18:28:40.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b920f01a9437416180bca614317ad601 : SUCCESS in 12m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b34bbc5e78044edb2616f3a0e328a59 : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/803127118c384d908a16ace34dd3f3cf : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1ac9fe736ca487d9bce3146b488bb83 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/749b6014e2a04f9eb24addcfd75f9eaf : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c214bb75ca394284a9c1097c49bed6e2 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0a89c101ad540f79f96960acb3bd683 : FAILURE in 6m 11s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d7dfc2d188a742e0bbb4c6b1c2798511 : SUCCESS in 5m 04s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c99cd16d7afb43548ece1e241d320a27 : SUCCESS in 8m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c9870fc2c07047b6924ab2adae8f8d08 : SUCCESS in 6m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/561a120b91f546c9b4525e9505194ffb : SUCCESS in 7m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/df2312a24d924b988acee0ba99401d7a : SUCCESS in 9m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/57ed52efe3c3463c9ebf1eed2a5545a3 : SUCCESS in 8m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e528c1bc6d75425f8b4a643f369a332c : SUCCESS in 2h 15m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9bbf08e892a34292a649825e4945fde7 : SUCCESS in 1h 29m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fb32208d2e7c4cceac5073070f69006d : SUCCESS in 36m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c875189dfd5f4dfb875cff7de8771b78 : SUCCESS in 53m 07s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0eb56698478f44308d2ba21e199d9079 : FAILURE in 16m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49b079a4e86c46188c3545f8be21cd1d : SUCCESS in 35m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3a448c559c7e4ca58e134768964fd640 : SUCCESS in 1h 43m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/74fef0d9ddd241c8b1a821e65abcb10e : SUCCESS in 1h 51m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f5d6f604306049d3a77de85e67795bd4 : SUCCESS in 1h 41m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/51567a88b8754adabd2863069e195bbc : SUCCESS in 2h 15m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2bdcebe6b25b4956b45d44b5bfbc67ed : TIMED_OUT in 2h 36m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/95334f094ac54a9d8a5139dbf4ba65ae : FAILURE in 13m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f6cf90a01856225ce1fb1934f3b842e45fda3c49","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-15 18:31:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"44d020baa78bd3020e6b245a032fda2d04dba6a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 20:44:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10490c1231fb4d90bda5e2ba9d356fb0 : SUCCESS in 14m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6959e444179a430f94a5b9b61c8964cb : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e45b0696d25d4e02892317caeba50663 : SUCCESS in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7a7792eda6a40a8a8f63e3eb0f67a6a : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4467373cfc624be380d8cde47d39a3eb : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5910fd2eb394cbdadc9b858d367f74a : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd00857b583b428bad9f9b4e90a89b0d : SUCCESS in 12m 00s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/99303245c1c540dbab0c7389c3ee86dd : SUCCESS in 5m 37s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8409d18cda6448ddaa5306c6148cb6a2 : SUCCESS in 8m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b18454ca175f4833911f42e9a163e3c6 : SUCCESS in 5m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/715353caf9be43a1b6e0bd5aa403db52 : SUCCESS in 8m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/970e9ba2ed09455cb90723f54fb61fab : SUCCESS in 8m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/92fbad98ae414dadaf24e446dc9545cb : SUCCESS in 10m 37s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53c7f770ba734ad08ac519eb1abb84c3 : SUCCESS in 2h 07m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/661ada89ceef433c99f55ec6e8aece1a : SUCCESS in 1h 27m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ff245979934042e09ed837ef0c75bf3b : SUCCESS in 33m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8c7cec51d946465fb2557c2d32aba8e9 : SUCCESS in 50m 10s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ded6cc92554f4a31a46d3ef827cfc826 : FAILURE in 15m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e3ee4295903547c7b579b058c5a32cb7 : SUCCESS in 32m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/77be746975734719a00abfdd2f14adc8 : SUCCESS in 1h 49m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/844b9f0579c845d098f4cd763f093f38 : SUCCESS in 1h 47m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0d83fd5efcef4fc1a95d36926289ade4 : SUCCESS in 1h 35m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/044de4c4c5a54890847a8cc5eb082f69 : SUCCESS in 1h 27m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7693b9333ac940b0a03e35a8494225bf : SUCCESS in 1h 36m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/ecb5e4417ca84f6d90b577cdd793cb15 : FAILURE in 13m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"309c12a450a3dee149090c0eee8c3ee1ab4d7a73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 20:45:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"234204a47ea6526a58e58e6a794c2687d10af23f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 23:25:17.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ee1b3bd60144c64a6080895f18991b0 : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b921a455eb541a8a832143811f419b3 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9e5f903d8aa47f193bfcd39ceb25a4f : SUCCESS in 7m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34c3e11140c04322921c8d7faa8c5614 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fbc37beeb1b46ac940792711e8f6399 : SUCCESS in 17m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/10ff1fa8c2fc45e3af62ca72c1f10066 : SUCCESS in 6m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5f60dacf05614a6c9fe12644785de548 : SUCCESS in 9m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/16aa375dcaa74ec1a535afbc8412ffd3 : SUCCESS in 1h 49m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4b9fe62e6bd4a8aa9cac2364c8a2bfb : TIMED_OUT in 2h 37m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9cae2069816c4927b24fd3d1948ef534 : SUCCESS in 1h 11m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e1b991a36c494fa5bf48c34e76c97f96 : SUCCESS in 52m 05s","accounts_in_message":[],"_revision_number":4},{"id":"3f61da0c1169900769d47a5d2e13ac966670857d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-15 23:32:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"24ec8c67738e9a6b976d1c6612b199d17b1ffc43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 02:42:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34f57660bc4b4ac1bbc7b2635d3d2c9d : SUCCESS in 10m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70ef91b4803e42d293ed88cdb81051ad : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d284314f10104a6b8cb68078790a25ff : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b01839cc1a194446b5c0c502a01a2229 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f642392a3e1442449632c5ad352170b1 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/133885bedf0d4d33bb2b4f4b047856c9 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af5a22837b5d4581ae3c448439c1cb5c : SUCCESS in 13m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/58f5d7d6b34a40c1851be5f9a43ffd33 : SUCCESS in 5m 47s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7cf23e9c118141ad8fee3bb835c9679e : SUCCESS in 7m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dc7115ded2b54e8ab4a25a4512ec01c1 : SUCCESS in 5m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a05a088afed543f58f4c8786f862ef30 : SUCCESS in 7m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/da265c59d52b48acaf984b8a96a33754 : SUCCESS in 7m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/30b211342b154b18a641d6d3c79ff2eb : SUCCESS in 8m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c4715210efea41d09c22d8a1a0c38853 : SUCCESS in 2h 14m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1978d7adf1ab44a8bdfc6fd21416ff50 : SUCCESS in 1h 48m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a943bba5de504bf182a00067cb486b0e : SUCCESS in 44m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/17bcc0cd95794f41b9e35c551790b566 : SUCCESS in 52m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2e24a5d4725b4ca0ac68882554b883ab : FAILURE in 16m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a3229fdf89f34f7e9adb89f7f6215dbd : SUCCESS in 31m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a582c25c5c4545189c3da577e5b20421 : SUCCESS in 1h 42m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/87a4a6e910d54d698db183c449e902b9 : SUCCESS in 2h 32m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/98e66c2fe80a47f9af05b624824e4641 : SUCCESS in 1h 37m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8f9aaee62bd64f01842935ab5bb3f413 : SUCCESS in 1h 37m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/988c4c8d6c7a40c2acce01e25674626d : SUCCESS in 1h 27m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/434aa94e7dcf4cb58a2ada8976003540 : FAILURE in 21m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1079d621e5374090d6a3572e98b1264b43928e11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 02:43:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ba6513d4f21608e5a52b7fff84bafbdf1a7a0992","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 04:49:43.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73313cfc8c7241229bb90c4fff7a23e0 : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c06594be0ab949c2b159e10584ce9e67 : SUCCESS in 9m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0832cd4b42704abf95a37784879c4f6c : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5230c032fd86466cbbb4b74d1ffb16af : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/481fc86b5a1c4445b6008b9abf05751f : SUCCESS in 13m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/726e08c4050f448ea0bdfbceb3055e86 : SUCCESS in 6m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b778e1922634423c8ed31943d6d4b47a : SUCCESS in 8m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c973cae0b0254f93b566e9290fb831ee : SUCCESS in 2h 04m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8447d2e2b09541bcb0d50ccdb2f23de3 : SUCCESS in 2h 00m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1d81e916da664095bb68c0c46533b503 : SUCCESS in 41m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b01688b6eb4473caf325a11bf1108d3 : SUCCESS in 1h 06m 16s","accounts_in_message":[],"_revision_number":4},{"id":"3a83f1988accaccd0ec307355447287920ac548c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 04:49:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ebd801342f231d859e31736e6585ee797bc34507","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 04:51:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c64312ecf9074fada17c9693fb14f336 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d590d6c7051a9d9941900efc85959e9158609e08","revisions":{"e132cf4ded7bb97d955ee6333b62c5c7d041e8fc":{"kind":"REWORK","_number":1,"created":"2022-06-08 14:47:28.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/46/845146/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/46/845146/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/845146/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/845146/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/845146/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/46/845146/1"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:45:51.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:47:26.000000000","tz":0},"subject":"Fix new pylint issues","message":"Fix new pylint issues\n\nThe new pylint release removed a few rule sets and added a check for generators. This patch updates Octavia for those changes.\n\nChange-Id: I2261342377e7acb1d2bc4fc742a351dfc66e4305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e132cf4ded7bb97d955ee6333b62c5c7d041e8fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e132cf4ded7bb97d955ee6333b62c5c7d041e8fc"}]},"branch":"refs/heads/master"},"23d55beb959eeb3c6d0559779889f8bc33e9bb80":{"kind":"REWORK","_number":2,"created":"2022-06-08 15:27:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/46/845146/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/46/845146/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/845146/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/845146/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/845146/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/46/845146/2"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:45:51.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 15:27:00.000000000","tz":0},"subject":"Fix new pylint issues","message":"Fix new pylint issues\n\nThe new pylint release removed a few rule sets and added a check for generators. This patch updates Octavia for those changes.\n\nChange-Id: I2261342377e7acb1d2bc4fc742a351dfc66e4305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23d55beb959eeb3c6d0559779889f8bc33e9bb80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23d55beb959eeb3c6d0559779889f8bc33e9bb80"}]},"branch":"refs/heads/master"},"650611879155a4ada6451e3c7c7c29769b257bba":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-06-10 08:38:46.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/845146/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/46/845146/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/845146/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/845146/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/845146/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/46/845146/3"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:45:51.000000000","tz":0},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-10 08:38:46.000000000","tz":0},"subject":"Fix new pylint issues","message":"Fix new pylint issues\n\nThe new pylint release removed a few rule sets and added a check for generators. This patch updates Octavia for those changes.\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/845293\nChange-Id: I2261342377e7acb1d2bc4fc742a351dfc66e4305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/650611879155a4ada6451e3c7c7c29769b257bba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/650611879155a4ada6451e3c7c7c29769b257bba"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d590d6c7051a9d9941900efc85959e9158609e08":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-06-10 10:20:08.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/845146/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/46/845146/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/845146/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/845146/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/845146/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/46/845146/4"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 14:45:51.000000000","tz":0},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-10 10:20:08.000000000","tz":0},"subject":"Fix new pylint issues","message":"Fix new pylint issues\n\nThe new pylint release removed a few rule sets and added a check for generators. This patch updates Octavia for those changes.\n\nChange-Id: I2261342377e7acb1d2bc4fc742a351dfc66e4305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d590d6c7051a9d9941900efc85959e9158609e08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d590d6c7051a9d9941900efc85959e9158609e08"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
