)]}'
{"id":"openstack%2Foctavia~845171","triplet_id":"openstack%2Foctavia~master~Ic953d5a63f630e556af0fde7096fe3d686eff06c","project":"openstack/octavia","branch":"master","topic":"fix-pyroute2-netns-mock","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-06-15 23:32:30.000000000","reason":"Dr. Jens Harbott (frickler) replied on the change"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-16 04:53:08.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-06-15 18:12:30.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"Ic953d5a63f630e556af0fde7096fe3d686eff06c","subject":"Fix pyroute2.netns mock in some tests","status":"MERGED","created":"2022-06-08 20:21:02.000000000","updated":"2022-06-16 04:54:25.000000000","submitted":"2022-06-16 04:53:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"845171-fix-pyroute2-netns-mock","meta_rev_id":"90aa912fee2095a081ec2adbd55fc8c60ee3342e","_number":845171,"virtual_id_number":845171,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-16 04:53:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-06-15 17:58:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-06-15 18:12:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-06-15 18:12:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-08 20:39:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-14 14:55:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-06-15 15:29:59.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-06-15 15:48:11.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-06-15 17:37:35.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"af53755fcfae235b1a52f3c78490dadb129a4a10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-08 20:21:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d157c62955ca1eef191b68ed663e02dd69c5766","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 20:39:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/162ab1f499a84a3e8c26a14eadc4d38a : SUCCESS in 13m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08a06df2a74d4ce9ada244034a318346 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c6a7e887af0432290da26fc6c6078f3 : FAILURE in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ce64ff38650410caec1ce200769a691 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fa7da3859a442dda54813666465597e : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6555120233f4f939ff104ac7cb12d95 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7dd11311dd54e3ebc6fc1468596b3c3 : FAILURE in 5m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/06d304f7509740d6a2558c044a2c7040 : SUCCESS in 5m 43s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2413039f6b7d4875bcc9558da800390b : SUCCESS in 8m 27s\n\nWarning:\n  Comments left for invalid file .pylintrc\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py","accounts_in_message":[],"_revision_number":1},{"id":"cee521d50e0261c21e4055a65dc81ab520c46ae2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-08 20:57:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f445fc9da926323547323ea3bd8c381f3917d1d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 21:09:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e230f3c3572846288924e517201be732 : SUCCESS in 10m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ac0035ea0514afc9017dad2b5fd96d0 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d84cead506ab478699e79c0684e3844c : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85d74846036b4baeb3a0b9f2fd8e7951 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fbde9cb1eac44d9b27a484e329b7f5a : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c60dbb11dd247efb5b510faf9ea3a67 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11fda5353e92417c95e954887dac5274 : FAILURE in 6m 08s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7de7761dfe2d45e58ba935815cb4378e : SUCCESS in 6m 25s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c5fec99768ac46568bfff2c70c74f006 : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":2},{"id":"771f4d95ce9ab8ec7122936db3eca1ff4bfc01cf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-14 14:55:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1180d9d006e5119a134c7335ab1d0c75b7d3ee37","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-15 15:35:42.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2b30befa0a5e263814d3e2a65ef2050b3a6da9d4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-15 15:38:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"27f6fbf3e6b3666e0d8a58cd39576d437e5f8f56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-15 15:39:27.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"59487ac093b97d6ae7400c4dc300447d2ba4c741","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-15 15:40:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6385dcfc3801e7433c79db5da8bc053acc8334c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-15 15:47:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a24efef81f7a04423516498a0ed7d072ed3031be","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-15 15:47:41.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b5c60e91351c8730321b6f67018e14a80fe4e944","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-15 15:48:11.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"2d877165a879fb962115ce28c302131131f43c51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 16:03:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbc85d2c17c34d38816764f83a8a4342 : SUCCESS in 12m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4cea33f918e430e8de93b5e2044effd : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35c8cd955782417891b9f106f9c004fb : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3177345a5b1c4ed48e25a159099597be : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9921bca871b44ea2a6e0c33a3cc4691c : SUCCESS in 6m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9601c6d84a846699bde904454ed1074 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61edddbb647f42c8a3c8eb74eb281bce : FAILURE in 6m 20s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/12f1667f08d84380b36b0a65fa8a5975 : SUCCESS in 5m 50s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/db9c92a3944e4b948d786ce497a9ae2b : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":4},{"id":"3734c2260ef2a1cbf92eb529df734a4df30e0f56","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-15 16:58:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f1bf65c6301c7c1ee343f3b81b87ad622210e466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 17:13:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00abfb053fc145c9a5355124d50c2f30 : SUCCESS in 12m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b89f46433e5a4f048ca3c8c7b7f57b88 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f8031d3110f479fa59e3e3bcf9c9d6c : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97f53e6f4e1b4b4198b266d36f137bda : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1eb452f2b9e34ff896e285fc02d4f462 : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/639e9932dc9a445cbe43b05a1f5f8ae4 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e31e45471d441afb3ff4bf7ff13d270 : SUCCESS in 12m 42s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2723461383144e3ca35ef2693fe75cda : SUCCESS in 6m 05s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/06d61380190040bea82e50daa7123ad0 : SUCCESS in 9m 08s","accounts_in_message":[],"_revision_number":4},{"id":"f874c71e45dcf3d0c89a5cdbb3ffd99a3894e35e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-15 17:37:35.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e2c9414e93be1b5fc1c098de3abcbb06dffeae32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-15 17:39:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8a970a8597121d381c3d2cf1dbcedd58fb6e933a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 17:57:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fad56bb53ade444497de71be4bf00726 : SUCCESS in 12m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a15cf26f8b7a48ae804d8e0d11a24cec : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c0b7e423a464e9e8aac5e20b6655c4b : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/588017037e6a45c8b0e88ec8c196907f : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef998e9e3add465f8f784196dbe2ebcd : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35a8e7cee3b147b99eec51180ae4a069 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee7b3a4a651448e994269fce5c805373 : SUCCESS in 12m 41s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e92bf53f8d0b488bb9b080e92b8ff186 : SUCCESS in 6m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fd68a8423dbd4db49b54efbcc2fff2a3 : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":5},{"id":"db047d357066108d9693cb8d0198cde4612fc945","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-15 17:58:06.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"18ac42c36d69bf6dd5602be11a22a3aee4c9d415","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-15 18:12:30.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1a9410fefda1b8506af0e77668d9377ee4b5d318","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 20:45:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ce18b606081eccaea463d3a31735aa45586d6e59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 23:25:19.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"d43b6343ee144008428af64a2efff9b983c6ea27","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-15 23:32:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a0eb4f88a9dbd4491f9db17a10a7113298af4c47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 23:50:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1046976e4514cc48a0e263e426de973 : SUCCESS in 9m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26f60dea86a44644801a3e13c9b2537a : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3470dbd756f4641a6c107e73a5016ff : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a647ddb2aaca41acacc28122c536b42c : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2657ecccaeeb4be5940418b1fed1c8e7 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c74b7ad8df314771b5f8afbdc29a2d9b : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a35687b2c0f543239859ca51258b5941 : SUCCESS in 11m 23s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/79f80a2f3c2f4eac97aa364a8d432574 : SUCCESS in 6m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fd639df1e49746f999e4e530b675f01b : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":5},{"id":"4e14caa221d4af0f1478751495c96bd76ce6c140","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 02:43:54.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"47a9d6d9f908b70338f9a767d429730f8e10aef0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 04:53:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbaefddd909e48b7b5d3741e16025006 : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbafc98040e94eef8b088d7d79aec9dc : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cadd36f8cdf949ff9323feff69f29ce2 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d158eb4efc904fc3a7e7a693ee5398d1 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d755facfcc39447e940d38766cac8891 : SUCCESS in 12m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d80176dec7a3412db52efbbacfb9e062 : SUCCESS in 2h 05m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96bcbbfbe09c45dcb3c903ed729ae682 : SUCCESS in 1h 43m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2bb8b6cc49db4fa2b83f40ccdd36daa9 : SUCCESS in 38m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2be4f31f05b74fd897735b8311594637 : SUCCESS in 50m 38s","accounts_in_message":[],"_revision_number":5},{"id":"6e2e6568f4b84f40636efcbc8dc23ba4cf9750b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 04:53:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"90aa912fee2095a081ec2adbd55fc8c60ee3342e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 04:54:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddd170e676144c38b8ccac7c12ffff09 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"66d7b6c7d106441aea2d606c5bcadf1a5d4958d2","revisions":{"f11a79eacdbe0e14180b190876d2bd82b008705f":{"kind":"REWORK","_number":1,"created":"2022-06-08 20:21:02.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/71/845171/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/845171/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/845171/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/845171/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/845171/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/845171/1"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-08 20:18:10.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-08 20:18:10.000000000","tz":120},"subject":"Fix pyroute2.netns mock in some tests","message":"Fix pyroute2.netns mock in some tests\n\nWith pyroute 0.6.11 the netns module import has been refactored so that\nit is no longer available as an attribute. Adapt our mocks accordingly\nso that the unit tests continue to work with both older versions and the\nnew one.\n\nChange-Id: Ic953d5a63f630e556af0fde7096fe3d686eff06c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f11a79eacdbe0e14180b190876d2bd82b008705f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f11a79eacdbe0e14180b190876d2bd82b008705f"}]},"branch":"refs/heads/master"},"3a3f4023ef759bf7ae8ab00113da34ab2ddc2d3c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-06-08 20:57:09.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/71/845171/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/845171/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/845171/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/845171/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/845171/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/845171/2"}}},"commit":{"parents":[{"commit":"23d55beb959eeb3c6d0559779889f8bc33e9bb80","subject":"Fix new pylint issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23d55beb959eeb3c6d0559779889f8bc33e9bb80"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-08 20:18:10.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-08 20:57:09.000000000","tz":0},"subject":"Fix pyroute2.netns mock in some tests","message":"Fix pyroute2.netns mock in some tests\n\nWith pyroute 0.6.11 the netns module import has been refactored so that\nit is no longer available as an attribute. Adapt our mocks accordingly\nso that the unit tests continue to work with both older versions and the\nnew one.\n\nChange-Id: Ic953d5a63f630e556af0fde7096fe3d686eff06c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a3f4023ef759bf7ae8ab00113da34ab2ddc2d3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a3f4023ef759bf7ae8ab00113da34ab2ddc2d3c"}]},"branch":"refs/heads/master","description":"Rebase"},"181aba081900bd2d1351d68a495d7ae0c05ed9c6":{"kind":"NO_CHANGE","_number":3,"created":"2022-06-15 15:39:27.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/71/845171/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/845171/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/845171/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/845171/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/845171/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/845171/3"}}},"commit":{"parents":[{"commit":"d590d6c7051a9d9941900efc85959e9158609e08","subject":"Fix new pylint issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d590d6c7051a9d9941900efc85959e9158609e08"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-08 20:18:10.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-06-15 15:39:27.000000000","tz":0},"subject":"Fix pyroute2.netns mock in some tests","message":"Fix pyroute2.netns mock in some tests\n\nWith pyroute 0.6.11 the netns module import has been refactored so that\nit is no longer available as an attribute. Adapt our mocks accordingly\nso that the unit tests continue to work with both older versions and the\nnew one.\n\nChange-Id: Ic953d5a63f630e556af0fde7096fe3d686eff06c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/181aba081900bd2d1351d68a495d7ae0c05ed9c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/181aba081900bd2d1351d68a495d7ae0c05ed9c6"}]},"branch":"refs/heads/master","description":"Rebase"},"3825c1f5c11b7653fa338b7a8a511636df7e3943":{"kind":"REWORK","_number":4,"created":"2022-06-15 15:47:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/845171/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/845171/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/845171/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/845171/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/845171/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/845171/4"}}},"commit":{"parents":[{"commit":"d590d6c7051a9d9941900efc85959e9158609e08","subject":"Fix new pylint issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d590d6c7051a9d9941900efc85959e9158609e08"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-08 20:18:10.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-15 15:45:39.000000000","tz":120},"subject":"Fix pyroute2.netns mock in some tests","message":"Fix pyroute2.netns mock in some tests\n\nWith pyroute 0.6.11 the netns module import has been refactored so that\nit is no longer available as an attribute. Adapt our mocks accordingly\nso that the unit tests continue to work with both older versions and the\nnew one.\n\nChange-Id: Ic953d5a63f630e556af0fde7096fe3d686eff06c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3825c1f5c11b7653fa338b7a8a511636df7e3943"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3825c1f5c11b7653fa338b7a8a511636df7e3943"}]},"branch":"refs/heads/master"},"66d7b6c7d106441aea2d606c5bcadf1a5d4958d2":{"kind":"REWORK","_number":5,"created":"2022-06-15 17:39:16.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/71/845171/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/845171/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/845171/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/845171/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/845171/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/845171/5"}}},"commit":{"parents":[{"commit":"d590d6c7051a9d9941900efc85959e9158609e08","subject":"Fix new pylint issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d590d6c7051a9d9941900efc85959e9158609e08"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-08 20:18:10.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-15 17:38:58.000000000","tz":0},"subject":"Fix pyroute2.netns mock in some tests","message":"Fix pyroute2.netns mock in some tests\n\nWith pyroute 0.6.11 the netns module import has been refactored so that\nit is no longer available as an attribute. Adapt our mocks accordingly\nso that the unit tests continue to work with both older versions and the\nnew one.\n\nChange-Id: Ic953d5a63f630e556af0fde7096fe3d686eff06c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66d7b6c7d106441aea2d606c5bcadf1a5d4958d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66d7b6c7d106441aea2d606c5bcadf1a5d4958d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
