)]}'
{"id":"openstack%2Foctavia~845294","triplet_id":"openstack%2Foctavia~master~I1177bc8bde0f6699bbdf83ca95c47f52714bc962","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-06-13 14:47:10.000000000","reason":"Change was abandoned"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-13 14:47:03.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I1177bc8bde0f6699bbdf83ca95c47f52714bc962","subject":"DNM: Testing taskflow fix for docs issue","status":"ABANDONED","created":"2022-06-09 18:08:16.000000000","updated":"2022-06-13 14:47:10.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"887d60279bdea9d89c529a363abddefc1a5c0f31","_number":845294,"virtual_id_number":845294,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-10 08:06:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-09 20:20:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf78e59767e8f8125ea308fd2f673476b1aa6acd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-09 18:08:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"106adc1b7c17f13ffb1a271bda7a56b449d6850c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 20:20:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f1fe5d2bee0246f5a154edb83af563f6 : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22bb37603fa94de9934aee163010512c : SUCCESS in 10m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d74605c7338f44a9b3bb50e37ce18caf : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1d879d4b00c47259526dc2779a15e06 : FAILURE in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5621f76bdc2443b9df6faa26de98f18 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e52b5729ddff4708b079a2bc83361157 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74059bfbb00845f3a3653bc3797c90ce : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3013cda32cbb4811b6776a56c0dfba85 : FAILURE in 5m 45s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/783cde57b2d84138a65832df56f187a3 : SUCCESS in 5m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c3d812e9eb9e44e3a8d856e49952511e : SUCCESS in 8m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e8f8ce57aa484e7c87ce9576301e49e7 : SUCCESS in 5m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4a0f0f065970402f8d44384d9628f44a : SUCCESS in 7m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e45c6a645d544e48b35915ea7cfbf780 : SUCCESS in 8m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/26630c5af28a472fb4aef4ae3dae1b18 : SUCCESS in 7m 21s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/25f91e29bd8d4fb1b3a0f57a1ea7f183 : SUCCESS in 2h 10m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a96d9749cc4482c9569b2753d9a281d : SUCCESS in 1h 23m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/21de2ccc67c6432e921b30e659fd0634 : SUCCESS in 34m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2fcdac222d2a479a865caeb53c5bf437 : FAILURE in 12m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6218a5c41703418e8074a2069536932c : SUCCESS in 32m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ded8dc7c4fc04fddab79174cde2d37f0 : SUCCESS in 1h 31m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/10f402410bbf4a81a9e74bff7122fe53 : SUCCESS in 2h 02m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4995c9f9b6ed47e5bbae0329842410ba : SUCCESS in 1h 53m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/129d5f39fc244a9e9393d909e920c076 : SUCCESS in 1h 34m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7a35750dddae4592be353c4f12d4fbd7 : SUCCESS in 1h 34m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/d08ef5f79d884bdba3e48575522ac1cd : FAILURE in 9m 13s (non-voting)\n\nWarning:\n  Comments left for invalid file .pylintrc\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py","accounts_in_message":[],"_revision_number":1},{"id":"55dd3d4caa8abb108f683f084456a986e756d7fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-09 21:48:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8b93e5b83e5c3e6fe7cb8161ee59cc7f81630b57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 21:48:57.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027templates\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n  #      - check-requirements\n  #      - periodic-stable-jobs-neutron\n  #      - openstack-cover-jobs\n  #      - openstack-lower-constraints-jobs\n  #      - openstack-python3-zed-jobs\n  #      - publish-openstack-docs-pti\n  #      - release-notes-jobs-python3\n  #      - octavia-tox-tips\n      check:\n        jobs:\n          - octavia-v2-dsvm-scenario-taskflow\n  #        - openstack-tox-pip-check-reqs:\n  #            irrelevant-files:\n  #              - ^.*\\.rst$\n  #              - ^api-ref/.*$\n  #              - ^doc/.*$\n  #              - ^etc/.*$\n  #              - ^octavia/tests/.*$\n  #              - ^releasenotes/.*$\n  #        - openstack-tox-functional-py38:\n  #            irrelevant-files:\n  #              - ^.*\\.rst$\n  #              - ^api-ref/.*$\n  #              - ^doc/.*$\n  #              - ^etc/.*$\n  #              - ^octavia/tests/unit/.*$\n  #              - ^releasenotes/.*$\n  #        - openstack-tox-functional-py39:\n  #            irrelevant-files:\n  #              - ^.*\\.rst$\n  #              - ^api-ref/.*$\n  #              - ^doc/.*$\n  #              - ^etc/.*$\n  #              - ^octavia/tests/unit/.*$\n  #              - ^releasenotes/.*$\n  #        - openstack-tox-functional-py310:\n  #            irrelevant-files:\n  #              - ^.*\\.rst$\n  #              - ^api-ref/.*$\n  #              - ^doc/.*$\n  #              - ^etc/.*$\n  #              - ^octavia/tests/unit/.*$\n  #              - ^releasenotes/.*$\n  #            voting: false\n  #        - octavia-v2-dsvm-noop-api:\n  #            irrelevant-files: \u0026irrelevant-files\n  #              - ^.*\\.rst$\n  #              - ^api-ref/.*$\n  #              - ^doc/.*$\n  #              - ^octavia/tests/.*$\n  #              - ^releasenotes/.*$\n  #        - octavia-v2-dsvm-scenario:\n  #            irrelevant-files: *irrelevant-files\n  #        - octavia-v2-dsvm-tls-barbican:\n  #            irrelevant-files: *irrelevant-files\n  #        - octavia-grenade:\n  #            irrelevant-files:\n  #              - ^.*\\.rst$\n  #              - ^api-ref/.*$\n  #              - ^doc/.*$\n  #              - ^octavia/tests/.*$\n  #              - ^releasenotes/.*$\n  #              - ^setup.cfg$\n  #              - ^tools/.*$\n  #              - ^(test-|)requirements.txt$\n  #              - ^tox.ini$\n  #        - octavia-v2-dsvm-tls-barbican-fips:\n  #            irrelevant-files: *irrelevant-files\n  #            voting: false\n  #        - octavia-v2-act-stdby-dsvm-scenario:\n  #            irrelevant-files: *irrelevant-files\n  #            voting: false\n  #        - octavia-v2-dsvm-cinder-amphora:\n  #            irrelevant-files: *irrelevant-files\n  #            voting: false\n  #        - octavia-v2-dsvm-scenario-two-node:\n  #            irrelevant-files: *irrelevant-files\n  #            voting: false\n  #        - octavia-v2-dsvm-scenario-ipv6-only:\n  #            irrelevant-files: *irrelevant-files\n  #            voting: false\n  #        - octavia-v2-dsvm-scenario-amphora-v2:\n  #            irrelevant-files: *irrelevant-files\n  #            voting: false\n  #        - octavia-v2-dsvm-scenario-amphora-v2-no-jobboard:\n  #            irrelevant-files: *irrelevant-files\n  #            voting: false\n  #        - octavia-v2-dsvm-scenario-centos-8-stream:\n  #            irrelevant-files: *irrelevant-files\n  #            voting: false\n      gate:\n        fail-fast: true\n        queue: octavia\n        jobs:\n          - openstack-tox-pip-check-reqs:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py38:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-tls-barbican\n          - octavia-grenade\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-focal:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos-8-stream:\n              branches: ^(?!stable/.*).*$\n          - octavia-amphora-image-build\n          - octavia-grenade-ffu\n  # Putting octavia-v2-dsvm-scenario-fips in periodic as centos 8 is too slow\n          - octavia-v2-dsvm-scenario-fips:\n              voting: false\n      experimental:\n        jobs:\n          - octavia-v2-dsvm-scenario-nftables\n          - octavia-v2-dsvm-noop-api-amphora-v2\n          - octavia-v2-dsvm-tls-barbican-amphora-v2\n          - octavia-v2-act-stdby-dsvm-scenario-amphora-v2\n          - octavia-grenade-amphora-v2\n          - octavia-v2-dsvm-cinder-amphora-v2\n          - octavia-v2-dsvm-scenario-two-node-amphora-v2\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":2},{"id":"62a442ff2587feb50e892df228b29baa7b2a7dd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-09 21:49:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"479332b7b64cfc13f977f87401c440f095eb1e6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 23:29:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- octavia-v2-dsvm-scenario-taskflow https://zuul.opendev.org/t/openstack/build/b45f6fa8c55a4ada8e40c8f9ded03636 : SUCCESS in 1h 37m 21s","accounts_in_message":[],"_revision_number":3},{"id":"f1eb82f0c195dea47c5d5826cba4b2bec6b331da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-10 06:36:40.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"7ebcc4e72b5791af8486864ade5cf8ead3429ba6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 08:06:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e1176422d44848b1431014523cd6fb : FAILURE in 5m 44s\n- octavia-v2-dsvm-scenario-taskflow https://zuul.opendev.org/t/openstack/build/c95e03929e2343989c2d5f1f144d7b9a : SUCCESS in 1h 28m 14s","accounts_in_message":[],"_revision_number":4},{"id":"fdb57f601657c9cb10849d74d187c95c1da182d5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-13 14:47:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"887d60279bdea9d89c529a363abddefc1a5c0f31","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-13 14:47:10.000000000","message":"Abandoned\n\nNo longer needed","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"95a3c3d758b25dafb23fc7afb457cac1a651e4eb","revisions":{"d6df5b6e07cee8a9e4f4ae10c3871dc0bfa5571a":{"kind":"REWORK","_number":1,"created":"2022-06-09 18:08:16.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/94/845294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/845294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/845294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/845294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/845294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/845294/1"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-09 18:07:41.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-09 18:08:14.000000000","tz":0},"subject":"DNM: Testing taskflow fix for docs issue","message":"DNM: Testing taskflow fix for docs issue\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/845293\nChange-Id: I1177bc8bde0f6699bbdf83ca95c47f52714bc962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6df5b6e07cee8a9e4f4ae10c3871dc0bfa5571a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6df5b6e07cee8a9e4f4ae10c3871dc0bfa5571a"}]},"branch":"refs/heads/master"},"b783cd5499477a2ecb7b982950ca51a09c5521cf":{"kind":"REWORK","_number":2,"created":"2022-06-09 21:48:19.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/94/845294/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/845294/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/845294/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/845294/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/845294/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/845294/2"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-09 18:07:41.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-09 21:48:13.000000000","tz":0},"subject":"DNM: Testing taskflow fix for docs issue","message":"DNM: Testing taskflow fix for docs issue\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/845293\nChange-Id: I1177bc8bde0f6699bbdf83ca95c47f52714bc962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b783cd5499477a2ecb7b982950ca51a09c5521cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b783cd5499477a2ecb7b982950ca51a09c5521cf"}]},"branch":"refs/heads/master"},"02dc5c97448132287a1c05dc88bbe13c4b8b1251":{"kind":"REWORK","_number":3,"created":"2022-06-09 21:49:39.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/94/845294/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/845294/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/845294/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/845294/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/845294/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/845294/3"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-09 18:07:41.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-09 21:49:32.000000000","tz":0},"subject":"DNM: Testing taskflow fix for docs issue","message":"DNM: Testing taskflow fix for docs issue\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/845293\nChange-Id: I1177bc8bde0f6699bbdf83ca95c47f52714bc962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02dc5c97448132287a1c05dc88bbe13c4b8b1251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02dc5c97448132287a1c05dc88bbe13c4b8b1251"}]},"branch":"refs/heads/master"},"95a3c3d758b25dafb23fc7afb457cac1a651e4eb":{"kind":"REWORK","_number":4,"created":"2022-06-10 06:36:40.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/94/845294/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/845294/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/845294/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/845294/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/845294/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/845294/4"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-09 18:07:41.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-06-10 06:36:40.000000000","tz":0},"subject":"DNM: Testing taskflow fix for docs issue","message":"DNM: Testing taskflow fix for docs issue\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/845293\nChange-Id: I1177bc8bde0f6699bbdf83ca95c47f52714bc962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95a3c3d758b25dafb23fc7afb457cac1a651e4eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95a3c3d758b25dafb23fc7afb457cac1a651e4eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
