)]}'
{"id":"openstack%2Foctavia~845812","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Id763dee0cc048d3d5fe601ce51097581d3c1c4e3","project":"openstack/octavia","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-06-16 07:25:38.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"Id763dee0cc048d3d5fe601ce51097581d3c1c4e3","subject":"Fix amphora build on stable/train","status":"MERGED","created":"2022-06-14 19:25:10.000000000","updated":"2022-06-21 14:13:18.000000000","submitted":"2022-06-21 14:11:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845812","meta_rev_id":"7e871571f124e336316b188807f1fa04b2d7cc7f","_number":845812,"virtual_id_number":845812,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-21 14:11:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2022-06-15 11:24:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-15 16:22:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-06-20 16:18:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-06-21 11:55:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-14 21:03:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-15 11:24:15.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2022-06-15 16:22:33.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-06-20 16:18:56.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"26960639d7452a4d3122b488ab9cf0b27f602363","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-14 19:25:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"593f2bcb54105050769299a26a1eeb243e5a69c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 21:03:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efd5862e021a44568a06fc33253b6224 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c8cb62d102c40efa0b5dc9dae8f0afe : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8fd2b9aa5f364ba18eae083d94fab390 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e356aa402a949f0af70a27925605af5 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d061df6650664fcfb7a4d8fe9a3d515e : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03ca0336d2e24026902c9afaa1d688cc : SUCCESS in 11m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/451ed25d43f042e0bfa9541097fe96ff : SUCCESS in 6m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cba64012d0c740508c0193daf880874a : SUCCESS in 7m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/66f36116cdbb4965a3ec66e8f773dd85 : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/20486088475d45858940a693a9b62360 : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/edeaa25bc7bf4283ad9eb653fa5b79e9 : SUCCESS in 1h 29m 42s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/11eed75cfab0477386a44c7fd506fbed : SUCCESS in 1h 05m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3ae4788270d47fb94b1292b08a526a6 : FAILURE in 20m 44s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f97138703c9b4b1d9245501909e354d2 : FAILURE in 22m 35s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1a354e294f4b42e3a157e13393a7cb37 : FAILURE in 13m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ef56d4e66c4d4614b701f721b869159f : RETRY_LIMIT in 8m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0969600d08d74ca68d79f5554209ddda : RETRY_LIMIT in 5m 54s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bc5997ed1696408982920a344b1bcb7b : FAILURE in 23m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e62d8b82529c410e8354627eac41c705 : FAILURE in 21m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71b7c6d093ea4c6f91e91b006076a51a : FAILURE in 19m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/79350acf3db94e398b2fb075ca93d2da : FAILURE in 19m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/733c54a12745475bb865e2bc3a0133f3 : FAILURE in 15m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4a151bbf4b6bcaa0ee106d7d225d29f0fc73f5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-15 05:56:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"879bf953dc8a9e9fb438248122fbab7d66db0a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 05:57:11.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch stable/train.  The error was:\n\n  expected a list for dictionary value @ data[\u0027templates\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        #- check-requirements\n        #- periodic-stable-jobs-neutron\n        #- openstack-cover-jobs\n        #- openstack-python-jobs\n        #- openstack-python3-train-jobs\n        #- publish-openstack-docs-pti\n        #- release-notes-jobs-python3\n        #- octavia-tox-tips\n      check:\n        jobs:\n          #- openstack-tox-functional:\n          #    irrelevant-files:\n          #      - ^.*\\.rst$\n          #      - ^api-ref/.*$\n          #      - ^doc/.*$\n          #      - ^etc/.*$\n          #      - ^releasenotes/.*$\n          #- openstack-tox-functional-py36:\n          #    irrelevant-files:\n          #      - ^.*\\.rst$\n          #      - ^api-ref/.*$\n          #      - ^doc/.*$\n          #      - ^etc/.*$\n          #      - ^releasenotes/.*$\n          #- octavia-v2-dsvm-noop-api:\n          #    nodeset: octavia-single-node-ubuntu-bionic\n          #- octavia-v2-dsvm-noop-py2-api:\n          #    nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-scenario:\n              nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-py2-scenario:\n              nodeset: octavia-single-node-ubuntu-bionic\n              voting: false\n          #- octavia-v2-dsvm-py2-scenario-centos-7:\n          #    voting: false\n          #- octavia-v2-dsvm-scenario-two-node:\n          #    nodeset: octavia-single-node-ubuntu-bionic\n          #    voting: false\n          #- octavia-v2-dsvm-py2-scenario-two-node:\n          #    nodeset: octavia-single-node-ubuntu-bionic\n          #    voting: false\n          #- octavia-v2-dsvm-tls-barbican:\n          #    nodeset: octavia-single-node-ubuntu-bionic\n          #    irrelevant-files:\n          #      - ^.*\\.rst$\n          #      - ^api-ref/.*$\n          #      - ^doc/.*$\n          #      - ^octavia/tests/unit/.*$\n          #      - ^releasenotes/.*$\n          #    voting: false\n          #- octavia-v2-dsvm-scenario-amphora-v2:\n          #    nodeset: octavia-single-node-ubuntu-bionic\n          #    irrelevant-files:\n          #      - ^.*\\.rst$\n          #      - ^api-ref/.*$\n          #      - ^doc/.*$\n          #      - ^octavia/tests/unit/.*$\n          #      - ^releasenotes/.*$\n          #    voting: false\n          #- octavia-v2-act-stdby-iptables-dsvm-scenario:\n          #    nodeset: octavia-single-node-ubuntu-bionic\n          #    voting: false\n          #- octavia-v2-act-stdby-iptables-dsvm-py2-scenario:\n          #    nodeset: octavia-single-node-ubuntu-bionic\n          #    voting: false\n          #- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7:\n          #    voting: false\n      gate:\n        fail-fast: true\n        queue: octavia\n        jobs:\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api:\n              nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-noop-py2-api:\n              nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-scenario:\n              nodeset: octavia-single-node-ubuntu-bionic\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-xenial:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-bionic:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos7:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-fedora28:\n              branches: ^(?!stable/.*).*$\n          - octavia-amphora-image-build\n\n  in \"openstack/octavia/zuul.d/projects.yaml@stable/train\", line 3, column 3","accounts_in_message":[],"_revision_number":2},{"id":"450b3ca807e6bba86e9882dd2ff31f0c5255d6cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-15 06:48:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7cfece2e636065ae3ee588152f3a6d6206922534","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 06:49:29.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch stable/train.  The error was:\n\n  expected a list for dictionary value @ data[\u0027templates\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates: {}\n        #- check-requirements\n        #- periodic-stable-jobs-neutron\n        #- openstack-cover-jobs\n        #- openstack-python-jobs\n        #- openstack-python3-train-jobs\n        #- publish-openstack-docs-pti\n        #- release-notes-jobs-python3\n        #- octavia-tox-tips\n      check:\n        jobs:\n          #- openstack-tox-functional:\n          #    irrelevant-files:\n          #      - ^.*\\.rst$\n          #      - ^api-ref/.*$\n          #      - ^doc/.*$\n          #      - ^etc/.*$\n          #      - ^releasenotes/.*$\n          #- openstack-tox-functional-py36:\n          #    irrelevant-files:\n          #      - ^.*\\.rst$\n          #      - ^api-ref/.*$\n          #      - ^doc/.*$\n          #      - ^etc/.*$\n          #      - ^releasenotes/.*$\n          #- octavia-v2-dsvm-noop-api:\n          #    nodeset: octavia-single-node-ubuntu-bionic\n          #- octavia-v2-dsvm-noop-py2-api:\n          #    nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-scenario:\n              nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-py2-scenario:\n              nodeset: octavia-single-node-ubuntu-bionic\n              voting: false\n          #- octavia-v2-dsvm-py2-scenario-centos-7:\n          #    voting: false\n          #- octavia-v2-dsvm-scenario-two-node:\n          #    nodeset: octavia-single-node-ubuntu-bionic\n          #    voting: false\n          #- octavia-v2-dsvm-py2-scenario-two-node:\n          #    nodeset: octavia-single-node-ubuntu-bionic\n          #    voting: false\n          #- octavia-v2-dsvm-tls-barbican:\n          #    nodeset: octavia-single-node-ubuntu-bionic\n          #    irrelevant-files:\n          #      - ^.*\\.rst$\n          #      - ^api-ref/.*$\n          #      - ^doc/.*$\n          #      - ^octavia/tests/unit/.*$\n          #      - ^releasenotes/.*$\n          #    voting: false\n          #- octavia-v2-dsvm-scenario-amphora-v2:\n          #    nodeset: octavia-single-node-ubuntu-bionic\n          #    irrelevant-files:\n          #      - ^.*\\.rst$\n          #      - ^api-ref/.*$\n          #      - ^doc/.*$\n          #      - ^octavia/tests/unit/.*$\n          #      - ^releasenotes/.*$\n          #    voting: false\n          #- octavia-v2-act-stdby-iptables-dsvm-scenario:\n          #    nodeset: octavia-single-node-ubuntu-bionic\n          #    voting: false\n          #- octavia-v2-act-stdby-iptables-dsvm-py2-scenario:\n          #    nodeset: octavia-single-node-ubuntu-bionic\n          #    voting: false\n          #- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7:\n          #    voting: false\n      gate:\n        fail-fast: true\n        queue: octavia\n        jobs:\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api:\n              nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-noop-py2-api:\n              nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-scenario:\n              nodeset: octavia-single-node-ubuntu-bionic\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-xenial:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-bionic:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos7:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-fedora28:\n              branches: ^(?!stable/.*).*$\n          - octavia-amphora-image-build\n\n  in \"openstack/octavia/zuul.d/projects.yaml@stable/train\", line 3, column 3","accounts_in_message":[],"_revision_number":3},{"id":"b59284a90dca2386d32bbede9292072b5ec5abd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-15 06:58:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4aab5da804de1c42fa2abc57dfa89ef5cce550c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 07:25:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1515c0831040464e9feb8d09a1f10b64 : FAILURE in 21m 42s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/39dabd45f31e4c49b0e48d04a49fcf6d : FAILURE in 25m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d944eb600e72cba13e033b0ddd51a115724956ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-15 08:11:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"08eb7e9451270d9357a1e8e8e0a67ca945414cd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 09:46:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07bb5afbe35a4a1daf19563a366eea28 : SUCCESS in 1h 32m 02s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/07488e04278d4c8191971ee9d42e19bc : SUCCESS in 1h 32m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1654b625ce108b58e8a7da12e162d471e1b0baf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-15 09:57:02.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0f02afa238cfba83539b58c7e6b919a5e32b7201","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2022-06-15 11:24:15.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f9cfa876f8eba5c552f2a935758bbd353728e1d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 12:13:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86888cfd627e434499d03819e06725f8 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/607606da42c04ccab932838360c949ac : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/03514c06936c4cfab0013e6134fef3d3 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8b8fa1763cc4e6bb5815a3a0319c93b : SUCCESS in 6m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a889f9acf854043bf8614c650d7ae83 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/515ee9500c7b48a7a2894af084b30831 : SUCCESS in 11m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bc01dc667e07445ab12b044725d797a0 : SUCCESS in 8m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bc7a081c0fcc40fea343fac94aa005f9 : SUCCESS in 9m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/983f6deaf45b48a8a7d38587d79c704d : SUCCESS in 7m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ae1b3e5fb97e4e1f8816428ea67fb7a8 : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6e5c8069f889402897fb6642f5c62db3 : SUCCESS in 1h 06m 50s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/725105e50dfd4e2a8bf5f2df7d4bc9ba : SUCCESS in 1h 11m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cab89121074d44f0b83f433d836df6ea : SUCCESS in 1h 27m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8e352470b0404bae86fc9397ef0e8ef2 : SUCCESS in 1h 44m 01s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/519e838414fe49c196e70d653bac0dae : FAILURE in 14m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0693cd31225a467c9a9f4717ec2bb3c2 : RETRY_LIMIT in 7m 22s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff6c7ecd91e542cf90dcbea6f3f78a0d : RETRY_LIMIT in 9m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ed6fdb7aa5264a3f96a5d6135bef7eb1 : SUCCESS in 38m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5245159a07c543f9bd6e7384fb78b178 : SUCCESS in 1h 27m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d83b109785548ffb30a1a68a41b6419 : FAILURE in 32m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f249ec6415a04df386cf9ff3de40fcb1 : FAILURE in 31m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d5e11676b6a34fb8866800b3a9c1d8f8 : FAILURE in 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1bcdee10f9f5d4a8721961fa6d5cde6fce5f4fcc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-15 16:22:33.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f05f8a79d73eb83a60947569a9bcf12a107930f2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-16 07:25:38.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"209adcb2ef7e061125f4d0575082ec74247b6096","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-20 08:14:50.000000000","message":"Removed Workflow-1 by \u003cGERRIT_ACCOUNT_29244\u003e\n","accounts_in_message":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"_revision_number":6},{"id":"8e2995a838f3f16eb7ccc44a6d7cf87b24dc7796","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-20 16:18:56.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"70e90f4c128874d182c4e9ac19b97100a0a86dc4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-21 11:55:55.000000000","message":"Patch Set 6: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"13af62e6acc39d62862d7d6c35c39c2fd2267ef5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 11:56:52.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"dc96efc2d09d35765d7920959f0fb6070948b8fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 14:11:35.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6f7c71ff5fe416eba4a5b48618033b1 : SUCCESS in 6m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6db6189081a4470b9edd9ad0e378296b : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a4ac9044eec4cbf9732d64d17a349bf : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b7e980ed6aa455f91d00c1748844737 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecc2a86fc1154e5c96d1171ee5fb21aa : SUCCESS in 12m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/44625921ed7e4836bc7ce9782d9298ee : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe07bf3e008949b8aed84b56afc16d18 : SUCCESS in 8m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bfe4987eb005448b98882f60abf1eac4 : SUCCESS in 1h 02m 47s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a5534f3e5756431597acc33e21100afe : SUCCESS in 1h 07m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a60f89c59836485c9c92fe840a48046a : SUCCESS in 1h 40m 34s","accounts_in_message":[],"_revision_number":6},{"id":"87b2a09d7e973aca1fba13461d7b3e106fe2ac2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 14:11:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"7e871571f124e336316b188807f1fa04b2d7cc7f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 14:13:18.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca78dbc0221b4e98891c2d9f578505d2 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"86b02d37dbdcd02cd3218c41f2bde79604dc3f60","revisions":{"3ba45f20c428a3cad2f28a9695b65182ff5b2745":{"kind":"REWORK","_number":1,"created":"2022-06-14 19:25:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/845812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/845812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/845812/1"}}},"commit":{"parents":[{"commit":"38d79f37204bb671216184ca9621c7050ae497eb","subject":"Reject invalid whitespace in HM url_path value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38d79f37204bb671216184ca9621c7050ae497eb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-14 19:20:56.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-14 19:25:02.000000000","tz":120},"subject":"WIP Testing DIB on stable/train","message":"WIP Testing DIB on stable/train\n\nChange-Id: Id763dee0cc048d3d5fe601ce51097581d3c1c4e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ba45f20c428a3cad2f28a9695b65182ff5b2745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ba45f20c428a3cad2f28a9695b65182ff5b2745"}]},"branch":"refs/heads/stable/train"},"fcfc1e1cbc94b0ad5762370d203ab6c32615bff4":{"kind":"REWORK","_number":2,"created":"2022-06-15 05:56:26.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/845812/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/845812/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/845812/2"}}},"commit":{"parents":[{"commit":"38d79f37204bb671216184ca9621c7050ae497eb","subject":"Reject invalid whitespace in HM url_path value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38d79f37204bb671216184ca9621c7050ae497eb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-14 19:20:56.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-15 05:56:04.000000000","tz":120},"subject":"WIP Testing DIB on stable/train","message":"WIP Testing DIB on stable/train\n\nChange-Id: Id763dee0cc048d3d5fe601ce51097581d3c1c4e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcfc1e1cbc94b0ad5762370d203ab6c32615bff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcfc1e1cbc94b0ad5762370d203ab6c32615bff4"}]},"branch":"refs/heads/stable/train"},"f459103dd6f8900fb3686f287da239928aff149f":{"kind":"REWORK","_number":3,"created":"2022-06-15 06:48:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/845812/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/845812/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/845812/3"}}},"commit":{"parents":[{"commit":"38d79f37204bb671216184ca9621c7050ae497eb","subject":"Reject invalid whitespace in HM url_path value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38d79f37204bb671216184ca9621c7050ae497eb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-14 19:20:56.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-15 06:48:53.000000000","tz":120},"subject":"WIP Testing DIB on stable/train","message":"WIP Testing DIB on stable/train\n\nChange-Id: Id763dee0cc048d3d5fe601ce51097581d3c1c4e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f459103dd6f8900fb3686f287da239928aff149f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f459103dd6f8900fb3686f287da239928aff149f"}]},"branch":"refs/heads/stable/train"},"5eace35b00e637169411b64335634aadf54f0496":{"kind":"REWORK","_number":4,"created":"2022-06-15 06:58:03.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/845812/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/845812/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/845812/4"}}},"commit":{"parents":[{"commit":"38d79f37204bb671216184ca9621c7050ae497eb","subject":"Reject invalid whitespace in HM url_path value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38d79f37204bb671216184ca9621c7050ae497eb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-14 19:20:56.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-15 06:57:57.000000000","tz":120},"subject":"WIP Testing DIB on stable/train","message":"WIP Testing DIB on stable/train\n\nChange-Id: Id763dee0cc048d3d5fe601ce51097581d3c1c4e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5eace35b00e637169411b64335634aadf54f0496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5eace35b00e637169411b64335634aadf54f0496"}]},"branch":"refs/heads/stable/train"},"792de35a8c7e8f9fbf55d9f52a8f777b7d8be472":{"kind":"REWORK","_number":5,"created":"2022-06-15 08:11:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/845812/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/845812/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/845812/5"}}},"commit":{"parents":[{"commit":"38d79f37204bb671216184ca9621c7050ae497eb","subject":"Reject invalid whitespace in HM url_path value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38d79f37204bb671216184ca9621c7050ae497eb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-14 19:20:56.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-15 08:11:10.000000000","tz":120},"subject":"WIP Testing DIB on stable/train","message":"WIP Testing DIB on stable/train\n\nChange-Id: Id763dee0cc048d3d5fe601ce51097581d3c1c4e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/792de35a8c7e8f9fbf55d9f52a8f777b7d8be472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/792de35a8c7e8f9fbf55d9f52a8f777b7d8be472"}]},"branch":"refs/heads/stable/train"},"86b02d37dbdcd02cd3218c41f2bde79604dc3f60":{"kind":"REWORK","_number":6,"created":"2022-06-15 09:57:02.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/845812/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/845812/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/845812/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/845812/6"}}},"commit":{"parents":[{"commit":"38d79f37204bb671216184ca9621c7050ae497eb","subject":"Reject invalid whitespace in HM url_path value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38d79f37204bb671216184ca9621c7050ae497eb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-14 19:20:56.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-15 09:54:48.000000000","tz":120},"subject":"Fix amphora build on stable/train","message":"Fix amphora build on stable/train\n\nOctavia stable/train installs diskimage-builder from the master branch,\nbut DIB has recently removed the support for py36.\nCheckout the last DIB commit that supports py36, and also use the\nrequirements from stable/train instead of master.\n\nChange-Id: Id763dee0cc048d3d5fe601ce51097581d3c1c4e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/86b02d37dbdcd02cd3218c41f2bde79604dc3f60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/86b02d37dbdcd02cd3218c41f2bde79604dc3f60"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
