)]}'
{"id":"openstack%2Foctavia~845979","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Id763dee0cc048d3d5fe601ce51097581d3c1c4e3","project":"openstack/octavia","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-06-21 11:55:46.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"13039":{"account":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"last_update":"2022-06-21 14:11:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id763dee0cc048d3d5fe601ce51097581d3c1c4e3","subject":"Fix amphora build on stable/ussuri","status":"MERGED","created":"2022-06-15 12:15:46.000000000","updated":"2022-06-21 14:13:30.000000000","submitted":"2022-06-21 14:11:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"845979","meta_rev_id":"2d17f7fbf496754ffbf2994d743ed650968ac2a5","_number":845979,"virtual_id_number":845979,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-21 14:11:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-16 16:04:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-06-21 11:54:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-06-21 11:55:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-15 13:43:04.000000000","updated_by":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"reviewer":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"state":"CC"},{"updated":"2022-06-15 14:31:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-15 16:17:15.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-06-21 11:54:26.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"fc9f52787b13e85cd0f91da6d0a29a36922f3c4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-15 12:15:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32973fb4f7d82fec544798878f6505e68d0df43d","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2022-06-15 13:43:04.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"24a1fa06adc58f90a0ac197b22aed2f5e66ad76c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 14:31:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13ea39aa5b74430b9f2891227278af9f : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36ab6dfe7c4045a69cd3278e4b3e840b : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1708d1d276f41ff8e92a936dde16d62 : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6df48e32a3b748d9859f57fd2aba7e12 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f36c33d0a64c4858bcd530f2b2dbfd65 : SUCCESS in 12m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5efffd4c0ba14dd3bfa3627ba17a6afd : SUCCESS in 7m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/725ddb3f437c4f5999fb597f93173481 : SUCCESS in 8m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6bbb12b6a41a46fab05e180752d6a61d : SUCCESS in 6m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ceba0241bb0d4673b5ff3d025ced7672 : SUCCESS in 1h 33m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e364df5ba8340a7bc5cc2e24eb82af6 : SUCCESS in 1h 33m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e13812e11d0e450ab5b76098315bdfc9 : SUCCESS in 35m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/516f36729b6641c1a2fe249d2690527b : SUCCESS in 48m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54c42941fa12410682b89edef6e8c17f : SUCCESS in 33m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7b3cf901f8524853a021710f4bd432d9 : SUCCESS in 1h 19m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/114590907e0544749274fca67992040d : RETRY_LIMIT in 8m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d0d9ce4a6344431992eeb6789afa863f : SUCCESS in 1h 42m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad709c9e9d3a064510a30af0fdaff392881b8363","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-15 16:17:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fd08ff6653e218d4aa56fdfc57cba64717e52657","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-15 16:19:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"627e43551549704b00f1f2a95f22c70b52626f56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-16 11:00:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b6bac9161e768a7ca3cde8d02f148072a315c3b7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-16 11:31:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7cda2687ea754455f256000634fc11acd6c8cc13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 13:09:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/679336bd6ea041898cb2b9e39980ea0f : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96e62dca352a4c5aa7cd533d2acdf493 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ac4359750f947b9929710469293c701 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95821e3df7ff4849a903e52db129c17e : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b38639cc7e9147318e30bfeb09cd893d : SUCCESS in 12m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e73f2ad8db8b4343a50ae3f914dac2e9 : SUCCESS in 7m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/316291ab717f444b857c685d2b8e10d9 : SUCCESS in 9m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/efba91c489424be1896ee6251b271f98 : SUCCESS in 7m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3436100d2a9f48059c7bac37a05aaedb : SUCCESS in 1h 29m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/854364daef494f3790843bda702b7bf5 : SUCCESS in 1h 42m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c6f3900ffcd74ff9a3683cbc40dfd28b : SUCCESS in 42m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c91e9566dab84456995ea82355f159e4 : SUCCESS in 44m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/567814103e3247aa9aada9280784ebcc : SUCCESS in 35m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2237f32a5f0e4470a1e11b2c930a33b6 : SUCCESS in 1h 56m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d298bf45750840afac07e06ff0cba02a : RETRY_LIMIT in 6m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d59e2a6e6b4c49eea95b2a625a7f0a4e : SUCCESS in 2h 01m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dae861130b4232655a0d6a0de0774b4009a7313c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-16 16:04:58.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4d00900ba7df17455be2545cbe45deee072c553","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-21 11:54:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"93688f02be60991064fd2833b54f9883f93e9bd2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-21 11:55:46.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1535b2cacb5945064510dbf7bc7f43608ebeed45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 11:56:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d6bbe62a3abab33b53b28f228f59905f632d1d59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 14:11:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5f9c3dda0bb4df98154b3ed230f4e80 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3df2e95dbbf54561982920dd51e26b38 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52d11fad2d2e4b4ba1b3d925ba41fafc : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd648c203bff444583dfdbca55f076cf : SUCCESS in 10m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0261b033ec6d422683e8074983c15ded : SUCCESS in 6m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e97e838efd844853b376be714f8468fa : SUCCESS in 1h 09m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7577bec1a9f84e41ba56b1034a53a310 : SUCCESS in 1h 43m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fff8cdae1b8a4c618c1c3d4d366db881 : SUCCESS in 43m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ae19b4b88f9f4a3db2cd5fd9fe428547 : SUCCESS in 33m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4857270e0b945a984a707b271cc5b13 : SUCCESS in 36m 48s","accounts_in_message":[],"_revision_number":2},{"id":"fa1fd2cf19ead7aa4ae1e8078725583c70b4549f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 14:11:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2d17f7fbf496754ffbf2994d743ed650968ac2a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 14:13:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0113c147e1934b518cc40b1c9dee1985 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"efeb4da03e31770246a8e715133ce9153561c8a5","revisions":{"633257bcd7b1e83894aa742e2e40e6d4e6768955":{"kind":"REWORK","_number":1,"created":"2022-06-15 12:15:46.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/79/845979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/845979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/845979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/845979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/845979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/845979/1"}}},"commit":{"parents":[{"commit":"5acee9e39f3fb1bad7e897f085c6fe5eb6826343","subject":"Reject invalid whitespace in HM url_path value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5acee9e39f3fb1bad7e897f085c6fe5eb6826343"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-14 19:20:56.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-15 12:15:40.000000000","tz":120},"subject":"Fix amphora build on stable/ussuri","message":"Fix amphora build on stable/ussuri\n\nOctavia stable/ussuri installs diskimage-builder from the master branch,\nbut DIB has recently removed the support for py36.\nCheckout the last DIB commit that supports py36, and also use the\nrequirements from stable/ussuri instead of master.\n\nChange-Id: Id763dee0cc048d3d5fe601ce51097581d3c1c4e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/633257bcd7b1e83894aa742e2e40e6d4e6768955"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/633257bcd7b1e83894aa742e2e40e6d4e6768955"}]},"branch":"refs/heads/stable/ussuri"},"efeb4da03e31770246a8e715133ce9153561c8a5":{"kind":"REWORK","_number":2,"created":"2022-06-16 11:00:53.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/79/845979/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/845979/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/845979/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/845979/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/845979/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/845979/2"}}},"commit":{"parents":[{"commit":"5acee9e39f3fb1bad7e897f085c6fe5eb6826343","subject":"Reject invalid whitespace in HM url_path value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5acee9e39f3fb1bad7e897f085c6fe5eb6826343"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-14 19:20:56.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-16 11:00:45.000000000","tz":120},"subject":"Fix amphora build on stable/ussuri","message":"Fix amphora build on stable/ussuri\n\nOctavia stable/ussuri installs diskimage-builder from the master branch,\nbut DIB has recently removed the support for py36.\nCheckout the last DIB commit that supports py36, and also use the\nrequirements from stable/ussuri instead of master.\n\nChange-Id: Id763dee0cc048d3d5fe601ce51097581d3c1c4e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/efeb4da03e31770246a8e715133ce9153561c8a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/efeb4da03e31770246a8e715133ce9153561c8a5"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
